)]}'
{"id":"openstack%2Fproject-config~730323","triplet_id":"openstack%2Fproject-config~master~I767359d48e04473fad0992b35c804ed511cfaa62","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I767359d48e04473fad0992b35c804ed511cfaa62","subject":"build and publish wheels for more distributions (x86-64)","status":"MERGED","created":"2020-05-22 13:31:24.000000000","updated":"2020-05-27 14:46:09.000000000","submitted":"2020-05-27 14:46:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730323-1590590769908-3f6755ef","meta_rev_id":"d6d25177d99532ba438d1915f47616571d97aa38","_number":730323,"virtual_id_number":730323,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 14:46:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-05-26 17:10:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 04:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":2,"date":"2020-05-27 14:31:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":1,"date":"2020-05-27 14:31:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-22 13:33:59.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-26 17:10:43.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-05-27 04:25:09.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-27 14:31:37.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-05-27 14:46:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb92477097c433402f4cd2058c8e5f8f137a8170","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-22 13:31:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8f2a8b19fed603b2faa4b352a7ccb87b64f9278","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-22 13:33:30.000000000","message":"Patch Set 1:\n\nrelease.yaml has afs_volume list which need to be filled with extra entries","accounts_in_message":[],"_revision_number":1},{"id":"126687f7dc6628703dedf4eb9b4729a706c729fc","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-22 13:36:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"56d3995a226bc8389d03188ae46761e792197b8c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-22 13:44:21.000000000","message":"Patch Set 2:\n\nAn AFS administrator will also need to manually allocate these new volumes.","accounts_in_message":[],"_revision_number":2},{"id":"bae6cc3c799288d3b2a119345c849776de270d62","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-22 13:44:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd1b5826eda05c92ace33070c8dad15ec9c9297b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-22 13:46:36.000000000","message":"Patch Set 1:\n\n\u003e An AFS administrator will also need to manually allocate these new\n \u003e volumes.\n\nNext patch will add second list with s/x64/arm64/ names.","accounts_in_message":[],"_revision_number":1},{"id":"47829375fc685a1558a47ea2ca73f69a6c3d932c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-22 13:52:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de725312aaf84e3db66689cdd8f495fa : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c65181f78ab4aa2f3a49bf88f70a4b06d4d37a72","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-26 17:10:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"df837b352d54e5450838c6efe47c137c143581b1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-27 00:13:28.000000000","message":"Patch Set 3: Code-Review-1\n\nI don\u0027t think the publish job knows where to write the buster wheel output into /afs/.openstack.org/mirror/wheels because the afs_slug isn\u0027t defined.\n\nhttps://opendev.org/openstack/project-config/src/branch/master/playbooks/publish/wheel-mirror.yaml#L18\n\neasy enough to fix but we should do that first","accounts_in_message":[],"_revision_number":3},{"id":"b41d118df2ddacf2fbc3ab44ea2ae9242afc7b8a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-27 04:25:09.000000000","message":"Patch Set 3: Code-Review+2\n\nwe should merge https://review.opendev.org/731009 first for this to work","accounts_in_message":[],"_revision_number":3},{"id":"81285b07f5f32e45d83de910b4009a94e06fc513","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-27 14:31:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThe AFS slug matching fix Ian mentions has merged now.","accounts_in_message":[],"_revision_number":3},{"id":"860976c4ff901adf76e58f8f0de02a462290c3bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 14:31:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0782cbe27855068e165a8e99cf160a021e75c244","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 14:46:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5bf58247db964f3387d378c8020db29d : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":3},{"id":"01cf8ec19a8959f487efceacbacf1a39589866a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 14:46:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3445673eeaf44e122a96c6210d7add595c3ecb5a","revisions":{"0d3ae2dd3036fde6eca1afeb2b08cfe86b3c9ba9":{"kind":"REWORK","_number":1,"created":"2020-05-22 13:31:24.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/730323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/730323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/730323/1"}}},"commit":{"parents":[{"commit":"a50f36da57df2afbe8c4b6aece38b233ad32271b","subject":"Replace old Ussuri cycle signing key with Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a50f36da57df2afbe8c4b6aece38b233ad32271b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:30:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:30:34.000000000","tz":120},"subject":"build and publish wheels for more distributions (x86-64)","message":"build and publish wheels for more distributions (x86-64)\n\n- CentOS 8\n- Debian \u0027buster\u0027\n- Ubuntu \u0027focal\u0027\n\nChange-Id: I767359d48e04473fad0992b35c804ed511cfaa62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0d3ae2dd3036fde6eca1afeb2b08cfe86b3c9ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0d3ae2dd3036fde6eca1afeb2b08cfe86b3c9ba9"}]},"branch":"refs/heads/master"},"e5121b9d78df9d5f0cce87a494316a077e9df356":{"kind":"REWORK","_number":2,"created":"2020-05-22 13:36:19.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/730323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/730323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/730323/2"}}},"commit":{"parents":[{"commit":"a50f36da57df2afbe8c4b6aece38b233ad32271b","subject":"Replace old Ussuri cycle signing key with Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a50f36da57df2afbe8c4b6aece38b233ad32271b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:30:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:36:09.000000000","tz":120},"subject":"build and publish wheels for more distributions (x86-64)","message":"build and publish wheels for more distributions (x86-64)\n\n- CentOS 8\n- Debian \u0027buster\u0027\n- Ubuntu \u0027focal\u0027\n\nChange-Id: I767359d48e04473fad0992b35c804ed511cfaa62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e5121b9d78df9d5f0cce87a494316a077e9df356"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e5121b9d78df9d5f0cce87a494316a077e9df356"}]},"branch":"refs/heads/master"},"3445673eeaf44e122a96c6210d7add595c3ecb5a":{"kind":"REWORK","_number":3,"created":"2020-05-22 13:44:38.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/23/730323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/730323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/730323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/730323/3"}}},"commit":{"parents":[{"commit":"a50f36da57df2afbe8c4b6aece38b233ad32271b","subject":"Replace old Ussuri cycle signing key with Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a50f36da57df2afbe8c4b6aece38b233ad32271b"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:30:34.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2020-05-22 13:44:17.000000000","tz":120},"subject":"build and publish wheels for more distributions (x86-64)","message":"build and publish wheels for more distributions (x86-64)\n\n- CentOS 8\n- Debian \u0027buster\u0027\n- Ubuntu \u0027focal\u0027\n\nChange-Id: I767359d48e04473fad0992b35c804ed511cfaa62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3445673eeaf44e122a96c6210d7add595c3ecb5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3445673eeaf44e122a96c6210d7add595c3ecb5a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
