)]}'
{"id":"openstack%2Fproject-config~730726","triplet_id":"openstack%2Fproject-config~master~I61f588912769274682cfadc1fca80da71d94d0a8","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I61f588912769274682cfadc1fca80da71d94d0a8","subject":"Add publish-to-pypi to ansible-config_template","status":"MERGED","created":"2020-05-26 08:08:02.000000000","updated":"2020-06-01 15:53:58.000000000","submitted":"2020-06-01 15:53:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730726-1591026838680-69ea3769","meta_rev_id":"38ec726bf0af2356ac89cdc968af7f52480774a4","_number":730726,"virtual_id_number":730726,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 15:53:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-06-01 15:45:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-05-29 05:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2020-05-28 19:08:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-06-01 15:45:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-05-29 04:58:20.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 19:08:45.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-05-29 05:06:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-01 15:45:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-06-01 15:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7993639ad8d93ee8b691d38e99524442492726df","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-26 08:08:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dfa51740b7372202ffdcede8e03b9ae99808b40","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-26 08:14:26.000000000","message":"Patch Set 1:\n\nAre you really publishing to pypi? Or do you just want release tarballs?","accounts_in_message":[],"_revision_number":1},{"id":"bc15711cd83c4aaddb7117abf37045797ef183e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 08:17:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ed774208146d4f7486a27294a39ac347 : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":1},{"id":"54c4ba6dd3239f5af7d11a01878cd87977fd5d01","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-26 08:32:12.000000000","message":"Patch Set 1: Workflow-1\n\n\u003e Are you really publishing to pypi? Or do you just want release\n \u003e tarballs?\n\nin RDO we rely only on tarballs, i proposed publish-to-pypi after seeing other ansible-projects used in TripleO following same, but yes good call, i will let Kevin answer this as he is managing it. For RDO case we can use tarball from github published tarball, and infact we can abandon this patch as https://review.opendev.org/#/c/730318/ already exist, so if publish is needed https://review.opendev.org/#/c/730318/ can be kept else revert.","accounts_in_message":[],"_revision_number":1},{"id":"c3cddbbca49dfef30edc1670f4bd5cf6cb5bc95e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-26 08:32:42.000000000","message":"Patch Set 1:\n\n\u003e \u003e Are you really publishing to pypi? Or do you just want release\n \u003e \u003e tarballs?\n \u003e \n \u003e in RDO we rely only on tarballs, i proposed publish-to-pypi after\n \u003e seeing other ansible-projects used in TripleO following same, but\n \u003e yes good call, i will let Kevin answer this as he is managing it.\n \u003e For RDO case we can use tarball from github published tarball, and\n \u003e infact we can abandon this patch as https://review.opendev.org/#/c/730318/\n \u003e already exist, so if publish is needed https://review.opendev.org/#/c/730318/\n \u003e can be kept else revert.\n\n@Kevin can u check this.","accounts_in_message":[],"_revision_number":1},{"id":"2b2ec45d9725697ef4d384d41a63b07a40a5cef4","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-05-28 19:08:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"93aa1f88c1bcf7de4f87f22d21a0b5e3d3b09a6c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2020-05-29 04:58:20.000000000","message":"Patch Set 1: -Workflow\n\nReady, Kevin has voted.","accounts_in_message":[],"_revision_number":1},{"id":"9f14c49354cdf0feccc9a74ae780904fbb21b54e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-29 05:06:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f801600eb628e75677cf587c844ec28a0e690b6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-06-01 15:45:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ef494b4f4a945b8872f3625e37bb18e8e77bde5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:45:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2773f5884c0ca33d6968771fea730fb353611c18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:53:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"38ec726bf0af2356ac89cdc968af7f52480774a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:53:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/28b3d8d1bfe64a9cb9525fef1d664bd2 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18dbb21c81a96161c0cf427ebed6eaa31e12e266","revisions":{"18dbb21c81a96161c0cf427ebed6eaa31e12e266":{"kind":"REWORK","_number":1,"created":"2020-05-26 08:08:02.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/26/730726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/730726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/730726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/730726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/730726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/730726/1"}}},"commit":{"parents":[{"commit":"7f10d908e622fbd8ac12a3a224d7047cef341150","subject":"Retire pabelanger as infra-root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f10d908e622fbd8ac12a3a224d7047cef341150"}]}],"author":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-26 08:03:03.000000000","tz":330},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2020-05-26 08:07:25.000000000","tz":330},"subject":"Add publish-to-pypi to ansible-config_template","message":"Add publish-to-pypi to ansible-config_template\n\nansible-config_template had missing publish jobs,\nthis patch adds it. The job added in project repo[1], but\nto get job run in already released tag adding template\nto project-config.\n\n[1] https://review.opendev.org/#/c/730318/\n\nChange-Id: I61f588912769274682cfadc1fca80da71d94d0a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18dbb21c81a96161c0cf427ebed6eaa31e12e266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18dbb21c81a96161c0cf427ebed6eaa31e12e266"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
