)]}'
{"id":"openstack%2Fproject-config~730818","triplet_id":"openstack%2Fproject-config~master~If03c299eea48aa73e0ef320d0f4903b2effb3ba1","project":"openstack/project-config","branch":"master","topic":"install-if-python","hashtags":[],"change_id":"If03c299eea48aa73e0ef320d0f4903b2effb3ba1","subject":"Fix sphinx playbook: install-if-python was renamed","status":"MERGED","created":"2020-05-26 13:43:34.000000000","updated":"2020-05-26 14:15:59.000000000","submitted":"2020-05-26 14:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730818-1590502559143-516f584a","meta_rev_id":"b251e1290689fbf6ded0b540cedece58a9764fec","_number":730818,"virtual_id_number":730818,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-26 14:15:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-05-26 13:44:26.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2020-05-26 13:44:26.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 13:44:26.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-05-26 14:15:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a502726ec47f9c7b894fa87565b1218a60a3f1c1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-26 13:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d090448e8710ca20c59e5fd308872b6af0f2cac1","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-05-26 13:44:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7365febcc205ec623dc8d6c38d3e59b1abaae978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/afc927694b574ca4a59855ec16bfcfb0 : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c11ceabccef1626f271f77fce12ac2460e48d3fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 14:00:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d52f702b076751c5bf9159c42e892e74a9b33e85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 14:15:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/18b631175cd54e8eb66cdd2eef41fda6 : SUCCESS in 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"55744a9a63eac9fba2d6275428bb6997182c464d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 14:15:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e607896c4185af443d317eb267a500175c1a7d4","revisions":{"8e607896c4185af443d317eb267a500175c1a7d4":{"kind":"REWORK","_number":1,"created":"2020-05-26 13:43:34.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/18/730818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/18/730818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/730818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/730818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/18/730818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/18/730818/1"}}},"commit":{"parents":[{"commit":"7f10d908e622fbd8ac12a3a224d7047cef341150","subject":"Retire pabelanger as infra-root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f10d908e622fbd8ac12a3a224d7047cef341150"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-26 13:42:45.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-26 13:42:45.000000000","tz":120},"subject":"Fix sphinx playbook: install-if-python was renamed","message":"Fix sphinx playbook: install-if-python was renamed\n\nThe new name is ensure-if-python, use it to unbreak the sphinx job.\n\nChange-Id: If03c299eea48aa73e0ef320d0f4903b2effb3ba1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e607896c4185af443d317eb267a500175c1a7d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e607896c4185af443d317eb267a500175c1a7d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
