)]}'
{"id":"openstack%2Fproject-config~730933","triplet_id":"openstack%2Fproject-config~master~I7965f296cdb1241c96293edd68e84e1d3d78b331","project":"openstack/project-config","branch":"master","topic":"update-pip-download-parsing","hashtags":[],"change_id":"I7965f296cdb1241c96293edd68e84e1d3d78b331","subject":"Update pip output parsing to fix wheel mirror builds","status":"MERGED","created":"2020-05-26 18:24:42.000000000","updated":"2020-05-27 01:03:40.000000000","submitted":"2020-05-27 01:03:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730933-1590541420013-69ea3769","meta_rev_id":"5383845581fd56d5d8d7d59fa04e6ac15639628c","_number":730933,"virtual_id_number":730933,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 01:03:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-05-26 22:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-27 00:57:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-27 00:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 22:38:00.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-05-27 00:57:01.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-05-27 01:03:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"829768eda870021349bda4b47d4903fb39129d13","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-26 18:24:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ce3e6f7a81e887fde704355c54c0ff82344664e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 18:36:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/51e8dc00775142bbbc6ae4e6194fe941 : SUCCESS in 9m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c3e35dad932bc53f1e9f90985350a30cc0cdf86f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-26 19:11:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c3fd62a9c6397989cb92cc155f27c2a664a5c25","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-26 19:24:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cda5aee417c296ead2a693da6c5faf5e5f838ebc","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-26 22:38:00.000000000","message":"Patch Set 1: Workflow-1\n\nAs we discussed in #opendev on Freenode, I\u0027ll update this with my suggested more liberal alternative to better future-proof the job (hopefully).","accounts_in_message":[],"_revision_number":1},{"id":"62d1c4a4ca99b89ac3c8cf276b0e56a0d66d71a8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-26 22:46:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"083d274e466984aafcca107ea29ed809446a129c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-05-26 22:46:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1fc67413ff5926d5de8a46f8fac7b6d01c18c8e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 22:54:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/955125bfd50b4dbd891ad39dc1c3629c : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":2},{"id":"7f21295ad7a4de3401a7e3815ab77bbc2652ac71","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-27 00:57:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e614e4c8373f849b8a0a4f6c3fb1ea46a46b6276","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 00:57:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e53f015deb370e36fd0cacdbff75e2fd0f33f4ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 01:03:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3fe69df060194a628666033dbeea297c : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"573ec484790906a89c012826781dba74182aa9f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 01:03:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b42840391a3ddb74122c35954a6f66238cd149e2","revisions":{"328ebd12539dae38f31bd08d978e6431fad59206":{"kind":"REWORK","_number":1,"created":"2020-05-26 18:24:42.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/33/730933/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/730933/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/730933/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/730933/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/730933/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/730933/1"}}},"commit":{"parents":[{"commit":"e750344180c0c401cfbf776e6fabd91f2ff9ba21","subject":"Merge \"Add base replication jobs for oslo-metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e750344180c0c401cfbf776e6fabd91f2ff9ba21"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-05-26 18:23:35.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-05-26 18:23:35.000000000","tz":-420},"subject":"Update pip output parsing to fix wheel mirror builds","message":"Update pip output parsing to fix wheel mirror builds\n\nIt seem that pip has updated its verbose output to say \"Downloading\n$filename\" rather than \"Downloading from url $URL\". We need to update\nour wheel mirror processing to handle this change so that it can\nproperly remove any pypi hosted wheels before publishing to our wheel\nmirrors.\n\nChange-Id: I7965f296cdb1241c96293edd68e84e1d3d78b331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328ebd12539dae38f31bd08d978e6431fad59206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/328ebd12539dae38f31bd08d978e6431fad59206"}]},"branch":"refs/heads/master"},"b42840391a3ddb74122c35954a6f66238cd149e2":{"kind":"REWORK","_number":2,"created":"2020-05-26 22:46:04.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/33/730933/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/730933/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/730933/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/730933/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/730933/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/730933/2"}}},"commit":{"parents":[{"commit":"e750344180c0c401cfbf776e6fabd91f2ff9ba21","subject":"Merge \"Add base replication jobs for oslo-metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e750344180c0c401cfbf776e6fabd91f2ff9ba21"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-05-26 18:23:35.000000000","tz":-420},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-26 22:44:44.000000000","tz":0},"subject":"Update pip output parsing to fix wheel mirror builds","message":"Update pip output parsing to fix wheel mirror builds\n\nIt seem that pip has updated its verbose output to say \"Downloading\n$filename\" rather than \"Downloading from url $URL\". We need to update\nour wheel mirror processing to handle this change so that it can\nproperly remove any pypi hosted wheels before publishing to our wheel\nmirrors. Make it more liberal so we match pretty much any line with\n\"Downloading\" and then eventually something which looks like a whl\nfile, just in case they keep fiddling with it.\n\nChange-Id: I7965f296cdb1241c96293edd68e84e1d3d78b331\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b42840391a3ddb74122c35954a6f66238cd149e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b42840391a3ddb74122c35954a6f66238cd149e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
