)]}'
{"id":"openstack%2Fproject-config~731380","triplet_id":"openstack%2Fproject-config~master~Ib54c2612c675ad6b161109ed2a281d440512bcc7","project":"openstack/project-config","branch":"master","topic":"centos7-arm-wheel","hashtags":[],"change_id":"Ib54c2612c675ad6b161109ed2a281d440512bcc7","subject":"wheel-build: remove centos-7-arm jobs","status":"MERGED","created":"2020-05-28 01:14:49.000000000","updated":"2020-05-28 02:57:36.000000000","submitted":"2020-05-28 02:57:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731380-1590634656890-ad19e745","meta_rev_id":"99d559904eade5386f9a6dff77fc295cfb4cc941","_number":731380,"virtual_id_number":731380,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 02:57:36.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 02:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-28 02:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 02:57:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49254ddce47b32d272872ea7219294614ab3a173","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 01:14:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efc265c6145160710509faa758defced60d81839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 01:23:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0d5bd6fe37a945d49b49c8264e27d253 : SUCCESS in 7m 09s","accounts_in_message":[],"_revision_number":1},{"id":"535fdd557ac274f16fa2caa4c2b7ca06a604aae4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-05-28 02:48:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\ni\u0027m taking the liberty of self-merging this because it just can not work, and at best is going to stop the publishing steps.  however, the jobs have other issues i\u0027m looking at in https://review.opendev.org/#/c/731387/2 and https://review.opendev.org/#/c/731381/","accounts_in_message":[],"_revision_number":1},{"id":"2d57d13a68ce98d704eacb36b21a559387f53ea4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 02:48:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db98927894772ea1adb755ab2a5224d4a72251ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 02:57:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ba0c5598f40ba1f67f314cb91e81b0af75eed386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 02:57:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/51c3eedd1b014cec96cd22e63aff71db : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4c7834daad62609165a103f9e016c659a3affac","revisions":{"e4c7834daad62609165a103f9e016c659a3affac":{"kind":"REWORK","_number":1,"created":"2020-05-28 01:14:49.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/80/731380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/80/731380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/731380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/731380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/80/731380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/80/731380/1"}}},"commit":{"parents":[{"commit":"ee746a6dc8fd9e8dd14c2a69d05ef5130c076d0d","subject":"Encrease vexxhost CA region presence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee746a6dc8fd9e8dd14c2a69d05ef5130c076d0d"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:12:54.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-28 01:13:56.000000000","tz":600},"subject":"wheel-build: remove centos-7-arm jobs","message":"wheel-build: remove centos-7-arm jobs\n\nFollowing on from Iebb1062269011664411691ef8fd55219afaccd1e; there are\nno arm64 CentOS 7 nodes, so this will not work right now and prevent\npublishing.  We can restore this if we feel we need it after making\nsuch nodes.\n\nChange-Id: Ib54c2612c675ad6b161109ed2a281d440512bcc7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e4c7834daad62609165a103f9e016c659a3affac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e4c7834daad62609165a103f9e016c659a3affac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
