)]}'
{"id":"openstack%2Fproject-config~731796","triplet_id":"openstack%2Fproject-config~master~Iffcd02aab14915c75ea75c2464acb0eb97c7f3a1","project":"openstack/project-config","branch":"master","topic":"deploy-ptgbot","hashtags":[],"change_id":"Iffcd02aab14915c75ea75c2464acb0eb97c7f3a1","subject":"Deploy ptgbot and puppet-ptgbot changes","status":"ABANDONED","created":"2020-05-29 16:04:30.000000000","updated":"2020-05-29 16:24:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"377cbb8828f426a6bd453eaf1e365a8b53e00df1","_number":731796,"virtual_id_number":731796,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-05-29 16:13:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-29 16:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d97ece8e870ad010a12ff487ea6677ea9d345fc0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-29 16:04:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"969be481c174634f9a60e03a9aaf5b95703eb228","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 16:13:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0fd25e34ecea49f19c8c9089e3cbbee2 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"251b0dd492f157747248e36c8b97f82110d0dff1","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-29 16:24:32.000000000","message":"Abandoned\n\nAfter discussion in IRC, I90ddc555ded0ac1d3134fd075d816155a475c6d2 is more future-proof even if not as immediate.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ff577d7c503563448fe90fdcf730662c98e8523","revisions":{"7ff577d7c503563448fe90fdcf730662c98e8523":{"kind":"REWORK","_number":1,"created":"2020-05-29 16:04:30.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/96/731796/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/96/731796/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/731796/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/731796/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/96/731796/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/96/731796/1"}}},"commit":{"parents":[{"commit":"457ded885a4a1c4534a7733f019e401b32d72bf0","subject":"Merge \"Add root_marker_dir for releasenotes publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/457ded885a4a1c4534a7733f019e401b32d72bf0"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-29 16:02:19.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-05-29 16:04:12.000000000","tz":0},"subject":"Deploy ptgbot and puppet-ptgbot changes","message":"Deploy ptgbot and puppet-ptgbot changes\n\nRun the service-eavesdrop playbook after changes merge for ptgbot or\nits puppet module.\n\nChange-Id: Iffcd02aab14915c75ea75c2464acb0eb97c7f3a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ff577d7c503563448fe90fdcf730662c98e8523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ff577d7c503563448fe90fdcf730662c98e8523"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
