)]}'
{"id":"openstack%2Fproject-config~732102","triplet_id":"openstack%2Fproject-config~master~I35e8cdd412006d10ba1bf114a649bdd86865daa0","project":"openstack/project-config","branch":"master","topic":"tasks/piotrowskim/kibana","hashtags":[],"change_id":"I35e8cdd412006d10ba1bf114a649bdd86865daa0","subject":"Adding elastic/kibana repository","status":"MERGED","created":"2020-06-01 07:46:48.000000000","updated":"2020-06-01 16:07:30.000000000","submitted":"2020-06-01 15:52:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732102-1591026742256-9d0d66de","meta_rev_id":"e01c191e53b9bcee7bc264e44890e0f49765175b","_number":732102,"virtual_id_number":732102,"owner":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 15:52:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-06-01 15:43:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-06-01 11:13:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-06-01 15:43:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 07:47:58.000000000","updated_by":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"reviewer":{"_account_id":26141,"name":"Adrian Czarnecki","email":"adrian.czarnecki@ts.fujitsu.com","username":"czarneckia"},"state":"REVIEWER"},{"updated":"2020-06-01 11:13:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-01 15:43:56.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-06-01 15:52:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f0e6f143aa7c49b1be8fe5ad812db7f7fcc35ab","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2020-06-01 07:46:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf8c091d7ddf8695306e31880abbea39a183a2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 07:55:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/52c1928c964c4c5fb29b62ddb5ad0f5e : SUCCESS in 7m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e6a03a441e7345f08812edc4d8273d40 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"aa6bbfcec2b776920925ca13cee362fae78f069e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-01 09:54:02.000000000","message":"Patch Set 1:\n\nplease explain *why* you add it in the commit message.\n\nWhere is this going to be used?","accounts_in_message":[],"_revision_number":1},{"id":"fe2fc8cc4ca8ef87c474f7520d0b27deb13ed094","author":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"date":"2020-06-01 10:28:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"93b3cfa92fc7eb7a968df9f656fe3a40357827e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 10:36:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0785e1607b934f959ed1accbd79aea98 : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/10ba57e8d0d34918bdf1109c54381efc : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3788c843522299ebb7563d0785c98505d9c97e38","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-01 11:13:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bc6ce875af81ad1d1c71fa2b554f76dd30a2056f","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2020-06-01 12:27:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"abda77f325de8ebebd6f6ad5ea02f285991bf603","author":{"_account_id":21922,"name":"Dobroslaw Zybort","email":"dobroslaw.zybort@ts.fujitsu.com","username":"dobrolaw.zybort"},"date":"2020-06-01 12:27:37.000000000","message":"Removed reviewer Dobroslaw Zybort with the following votes:\n\n* Code-Review+1 by Dobroslaw Zybort \u003cdobroslaw.zybort@ts.fujitsu.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"f98c9a957bb203ad760a4b5d6b829ff637956b49","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-06-01 15:43:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"06508fba8cd2bd2b85c0c7e97a02c97da6ba81c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:44:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1c819dc11041a044b9724b725f24079585d445d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:52:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2796bbce2d0fdc55892f4e253b286d147a7633b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 15:52:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/26a8f2754e4745968282f611b3c15fa1 : SUCCESS in 6m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/cc27588e3652497abfe7fba0189e8882 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e3914bf388213122c69d3f188ee3a24d518a1181","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 16:07:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/d382afb0700a46e7b07e689e3c457141 : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16f5d175fc32f81fb181c4221d88090ace33721f","revisions":{"a286eb19a8388491d8b70e86c5c2788e846ed05a":{"kind":"REWORK","_number":1,"created":"2020-06-01 07:46:48.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/02/732102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/732102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/732102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/732102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/732102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/732102/1"}}},"commit":{"parents":[{"commit":"457ded885a4a1c4534a7733f019e401b32d72bf0","subject":"Merge \"Add root_marker_dir for releasenotes publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/457ded885a4a1c4534a7733f019e401b32d72bf0"}]}],"author":{"name":"Michal Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2020-06-01 07:45:00.000000000","tz":120},"committer":{"name":"Michal Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2020-06-01 07:45:00.000000000","tz":120},"subject":"Adding elastic/kibana repository","message":"Adding elastic/kibana repository\n\nChange-Id: I35e8cdd412006d10ba1bf114a649bdd86865daa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a286eb19a8388491d8b70e86c5c2788e846ed05a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a286eb19a8388491d8b70e86c5c2788e846ed05a"}]},"branch":"refs/heads/master"},"16f5d175fc32f81fb181c4221d88090ace33721f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-01 10:28:16.000000000","uploader":{"_account_id":29909,"name":"Michał Piotrowski","email":"piotrowskimichalwfis@gmail.com","username":"tatarinho"},"ref":"refs/changes/02/732102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/02/732102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/732102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/732102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/02/732102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/02/732102/2"}}},"commit":{"parents":[{"commit":"457ded885a4a1c4534a7733f019e401b32d72bf0","subject":"Merge \"Add root_marker_dir for releasenotes publish\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/457ded885a4a1c4534a7733f019e401b32d72bf0"}]}],"author":{"name":"Michal Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2020-06-01 07:45:00.000000000","tz":120},"committer":{"name":"Michal Piotrowski","email":"michal.piotrowski@ts.fujitsu.com","date":"2020-06-01 10:23:35.000000000","tz":120},"subject":"Adding elastic/kibana repository","message":"Adding elastic/kibana repository\n\nThe purpose of this commit is to add possibility to get elastic/kibana\nrepository in preferable way. In monasca-kibana-plugin repository\nis needed to create Zuul job which checks if plugin builds.\n\nChange-Id: I35e8cdd412006d10ba1bf114a649bdd86865daa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16f5d175fc32f81fb181c4221d88090ace33721f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/16f5d175fc32f81fb181c4221d88090ace33721f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
