)]}'
{"id":"openstack%2Fproject-config~733182","triplet_id":"openstack%2Fproject-config~master~I9b9d398807d08de3a3fe7529ed051c3b95b7fa46","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I9b9d398807d08de3a3fe7529ed051c3b95b7fa46","subject":"Retire puppet-congress project - step 1 end project gating","status":"ABANDONED","created":"2020-06-03 15:41:32.000000000","updated":"2020-06-22 16:05:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4deb8c25254d670dd71e4c25a653b77d7e225060","_number":733182,"virtual_id_number":733182,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-03 15:48:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-06-09 13:26:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":-1,"date":"2020-06-05 19:02:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"date":"2020-06-09 13:25:16.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"date":"2020-06-09 13:25:04.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 15:48:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-05 19:02:31.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-09 13:25:04.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-06-09 13:25:16.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"9582170fe67593531fb30b2c0ab470372e4aa6b8","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-03 15:41:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"936a60ad552ec526b13b902f653cf76403419ba8","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-03 15:43:51.000000000","message":"Patch Set 1:\n\nplease push all changes together, see https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#retiring-a-project\n\nAlso, this change needs a link to the announcement.","accounts_in_message":[],"_revision_number":1},{"id":"0c05e8d647ee809fc16aeb315c99913fab8becea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 15:48:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/671d102b447d49f396cd34baa72ff043 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c944ac002516779ce1cf819475bd3562e4367f3c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-05 19:02:31.000000000","message":"Patch Set 1: Workflow-1\n\nwaiting for announcement and all changes pushed - plus PTL +1.","accounts_in_message":[],"_revision_number":1},{"id":"b7a7df014692c4b531af8f7b7a0c1669892cf825","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 13:26:59.000000000","message":"Patch Set 1: Code-Review-1\n\nPlease move this forward properly following the instructions at https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#retiring-a-project or abandon.\n\nThis misses:\n* needed-by/depends-on any other changes in commit message\n* an announcement on the mailing list, best with link in commit message\n* PTL approval","accounts_in_message":[],"_revision_number":1},{"id":"4deb8c25254d670dd71e4c25a653b77d7e225060","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-22 16:05:19.000000000","message":"Abandoned\n\na duplicate merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a898b7329face35ed738f87ddebe9b7bfeee9456","revisions":{"a898b7329face35ed738f87ddebe9b7bfeee9456":{"kind":"REWORK","_number":1,"created":"2020-06-03 15:41:32.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/82/733182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/82/733182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/733182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/733182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/82/733182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/82/733182/1"}}},"commit":{"parents":[{"commit":"03cdd9e1b10d0777851576c2524e3b72bbd3b0f9","subject":"Update Octavia grafana dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03cdd9e1b10d0777851576c2524e3b72bbd3b0f9"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-03 15:37:04.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-03 15:37:04.000000000","tz":480},"subject":"Retire puppet-congress project - step 1 end project gating","message":"Retire puppet-congress project - step 1 end project gating\n\nmore info: https://review.opendev.org/721742\n\nChange-Id: I9b9d398807d08de3a3fe7529ed051c3b95b7fa46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a898b7329face35ed738f87ddebe9b7bfeee9456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a898b7329face35ed738f87ddebe9b7bfeee9456"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
