)]}'
{"id":"openstack%2Fproject-config~733283","triplet_id":"openstack%2Fproject-config~master~I9c2b4fdbc48d5cbcb026a362ed49bd65b0afe09f","project":"openstack/project-config","branch":"master","topic":"review-difficulty","hashtags":[],"change_id":"I9c2b4fdbc48d5cbcb026a362ed49bd65b0afe09f","subject":"Propose Review-Difficulty label for Kayobe","status":"ABANDONED","created":"2020-06-03 17:15:24.000000000","updated":"2020-07-07 19:55:37.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a0a9adc8c8e5b217362789e6d3f3c8927ab1aab4","_number":733283,"virtual_id_number":733283,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-03 17:39:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":-1,"date":"2020-06-04 09:35:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-06-03 17:30:01.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 17:16:06.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2020-06-03 17:16:06.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-03 17:16:06.000000000","updated_by":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"reviewer":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"state":"REVIEWER"},{"updated":"2020-06-03 17:16:06.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-03 17:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-04 09:35:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1d77a44caba794f3f95d5f90ed490354f259ad68","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-03 17:15:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"870d4696cc77f6ff9476e159142f35e90b02beca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-03 17:16:01.000000000","message":"Patch Set 1: Workflow-1\n\nNeed discussion with other Kayobe cores first","accounts_in_message":[],"_revision_number":1},{"id":"e682cfbadef55a969357249ba13ad37a168c09b6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-03 17:16:29.000000000","message":"Patch Set 1:\n\n@kayobe-core reviewers: Would this be useful?","accounts_in_message":[],"_revision_number":1},{"id":"2a2f26ccb044cb190744b6ec15f59b31fbe5708f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:24:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2a525d8ca55f4b49bdd02bd894d84bd0 : SUCCESS in 7m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d6341d5c76e444ff8044bcb266a06dc3 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3cf73892dc951dcc366471bcf6961a138e1509da","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-03 17:29:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"717ad2c33349c55b4c489a5b0930faa60878f6a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-03 17:30:01.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8a34a91b9cdc0db58c28f2fc98d3353c8711980d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 17:39:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/83f1f0718068408f9d369fe74f5f34dd : SUCCESS in 6m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/aed2f10f97434aa9aca68105bf38b716 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1219eba10637479bc4a4b38ccaf9ba97e4946e9e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 09:15:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)\n\nMakes sense to me.","accounts_in_message":[],"_revision_number":2},{"id":"46fefba9a99af7d38db820825ff3cb4444f76e5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 09:35:36.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"471d66ec93bc2cbecc3ab6d108eb27af17c2abee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-07 19:55:37.000000000","message":"Abandoned\n\nAbandoning for now, we already have many Gerrit labels…","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"acab213f741e5968f4ee638581531b2871020196","revisions":{"50bbc92c41728ebb8387c5cd5f02d0e091e27f5b":{"kind":"REWORK","_number":1,"created":"2020-06-03 17:15:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/733283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/733283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/733283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/733283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/733283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/733283/1"}}},"commit":{"parents":[{"commit":"03cdd9e1b10d0777851576c2524e3b72bbd3b0f9","subject":"Update Octavia grafana dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03cdd9e1b10d0777851576c2524e3b72bbd3b0f9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-03 17:14:53.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-03 17:14:53.000000000","tz":120},"subject":"Propose Review-Difficulty label for Kayobe","message":"Propose Review-Difficulty label for Kayobe\n\nWhen very limited time is available for reviews, it can be difficult to\nchoose from a long list of open changes. I would like to propose the\nReview-Difficulty label (initially for Kayobe), to flag easy or trivial\nchanges that can be reviewed in a few minutes or seconds.\n\nThe goal is to make our review queue shorter by focusing core reviewers\nwhen they have very limited time available. Of course Review-Priority\nshould take precedence when more time is available for reviews.\n\nChange-Id: I9c2b4fdbc48d5cbcb026a362ed49bd65b0afe09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50bbc92c41728ebb8387c5cd5f02d0e091e27f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/50bbc92c41728ebb8387c5cd5f02d0e091e27f5b"}]},"branch":"refs/heads/master"},"acab213f741e5968f4ee638581531b2871020196":{"kind":"REWORK","_number":2,"created":"2020-06-03 17:29:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/733283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/733283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/733283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/733283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/733283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/733283/2"}}},"commit":{"parents":[{"commit":"03cdd9e1b10d0777851576c2524e3b72bbd3b0f9","subject":"Update Octavia grafana dashboard","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/03cdd9e1b10d0777851576c2524e3b72bbd3b0f9"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-03 17:29:01.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-03 17:29:01.000000000","tz":120},"subject":"Propose Review-Difficulty label for Kayobe","message":"Propose Review-Difficulty label for Kayobe\n\nWhen very limited time is available for reviews, it can be difficult to\nchoose from a long list of open changes. I would like to propose the\nReview-Difficulty label (initially for Kayobe), to flag easy or trivial\nchanges that can be reviewed in a few minutes or seconds.\n\nThe goal is to make our review queue shorter by focusing core reviewers\nwhen they have very limited time available. Of course Review-Priority\nshould take precedence when more time is available for reviews.\n\nChange-Id: I9c2b4fdbc48d5cbcb026a362ed49bd65b0afe09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/acab213f741e5968f4ee638581531b2871020196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/acab213f741e5968f4ee638581531b2871020196"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
