)]}'
{"id":"openstack%2Fproject-config~734393","triplet_id":"openstack%2Fproject-config~master~I983d23c33f7780e5708aa728c829c3262fc99ea0","project":"openstack/project-config","branch":"master","topic":"explicit-python3","hashtags":[],"change_id":"I983d23c33f7780e5708aa728c829c3262fc99ea0","subject":"Be more explicit about using python3 to run tools/","status":"MERGED","created":"2020-06-08 23:42:09.000000000","updated":"2020-06-10 00:05:20.000000000","submitted":"2020-06-10 00:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734393-1591747520658-b1fdb8ea","meta_rev_id":"3aec10314971f482ed7e30bf80891c7d371f9508","_number":734393,"virtual_id_number":734393,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 00:05:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-06-09 05:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-09 23:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-09 23:56:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 05:59:24.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-09 23:56:36.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-06-10 00:05:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66c0ddc16a495fcf7b61a7cf03f77a381133a26a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-06-08 23:42:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfc1ed4bea64ce2a74ed6bed059e9525ca1e5e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:49:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2ed9373229bd4bc8b8e4691ac6c04eb4 : SUCCESS in 7m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0fa4230fe0954c69b0e90714f7a094da : SUCCESS in 4m 43s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/af3a12bd0398440196083032e5b3c323 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"867820324e719044c0d094ea19bc71f7c69d8b98","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 05:59:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"527aedd16719c4896fdd7dc4ffeec94c77bc174a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-06-09 21:51:46.000000000","message":"Topic changed from python3-tools to explicit-python3","accounts_in_message":[],"_revision_number":1},{"id":"9bdcd7e6ce714e400cefc8ac6b09b43a9de6efa7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-09 23:56:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a9d3bcd0523aad902264f8c66dd981e519dc4372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 23:56:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"854c9fcb52b41b34221656e8765ae353543443f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:05:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/04eaf2646fbd4bccaa99670e185e383e : SUCCESS in 6m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2dbfb6c88495449b8c2a0aedcfa34ad3 : SUCCESS in 5m 17s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/4ddd04b0da8548fbb410ceaae60aa88b : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"4f25a07d39169acdffc25aecf93aec1b27f75253","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 00:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"180cc6aedacaa09db8d9c8faa8c6fe91f752be7e","revisions":{"180cc6aedacaa09db8d9c8faa8c6fe91f752be7e":{"kind":"REWORK","_number":1,"created":"2020-06-08 23:42:09.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/93/734393/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/734393/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/734393/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/734393/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/734393/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/734393/1"}}},"commit":{"parents":[{"commit":"b59a59da354233af247b611f616894484616cd10","subject":"Use python 3 venv for generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b59a59da354233af247b611f616894484616cd10"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-06-08 23:40:44.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2020-06-08 23:40:44.000000000","tz":-420},"subject":"Be more explicit about using python3 to run tools/","message":"Be more explicit about using python3 to run tools/\n\nWe have python scripts in the tools/ dir the vast majority of which we\nrun regularly with python3 via our python3 default basepython in tox.\nHowever, most of these use a `python` shebang line which can be\nconfusing as to whether or not these scripts run under python3 or not.\n\nTo make this more clear set them to python3. I\u0027ve confirmed the scripts\nrunning under tox are happy with these changes. For the ones that don\u0027t\nrun under tox I\u0027ve done a quick review and they look happy too.\n\nChange-Id: I983d23c33f7780e5708aa728c829c3262fc99ea0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/180cc6aedacaa09db8d9c8faa8c6fe91f752be7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/180cc6aedacaa09db8d9c8faa8c6fe91f752be7e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
