)]}'
{"id":"openstack%2Fproject-config~734640","triplet_id":"openstack%2Fproject-config~master~Idaf5c54c101f3af0dcbf42e23caae236f5731a6a","project":"openstack/project-config","branch":"master","topic":"paunch/retire","hashtags":[],"change_id":"Idaf5c54c101f3af0dcbf42e23caae236f5731a6a","subject":"Deprecate Paunch","status":"MERGED","created":"2020-06-09 16:14:19.000000000","updated":"2020-06-12 15:07:09.000000000","submitted":"2020-06-12 15:07:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734640-1591974429729-00e32a87","meta_rev_id":"495586ce39279cdb432333b23b04ad06059074ac","_number":734640,"virtual_id_number":734640,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-12 15:07:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-06-10 17:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-06-10 18:51:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2020-06-12 14:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2020-06-12 14:59:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 15:22:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-06-10 17:55:39.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-10 18:51:04.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-06-12 14:59:02.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-06-12 15:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"38c770ab8d2a79b387fcede2b559b81f20f65ab3","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-09 16:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed4f84817934e1c59db2159497852c63b43a5d4c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 16:16:29.000000000","message":"Patch Set 1:\n\ndid you sent an announcement? Please add a link here to it.\n\nAlso, link all changes via depends-On/Needed-By.","accounts_in_message":[],"_revision_number":1},{"id":"a09ba7d0eacb649198bdb07c496360159777ee77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 16:26:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/da9b800d75d64f8d80933938812d4ae6 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0d76b6b011d714e23b9e462dbe1e96cbe66e5c23","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-09 16:57:55.000000000","message":"Patch Set 1: Workflow-1\n\nSee previous comment, I\u0027d like to see the whole 4 changes submitted, the PTL +1 and a mailing list announcement linked.","accounts_in_message":[],"_revision_number":1},{"id":"e1715424e296255c6c7f9d8161b4266c59d83c79","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 13:19:42.000000000","message":"Patch Set 1: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"095e6440f795c507c82653644142a7d2c65ffb74","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 13:27:25.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"057f55952ec246783c29fc9b2eb0fbe85aacd525","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 13:30:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"caef7b8fb7c76a8787d15d34fe16c19db6e5e631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 13:37:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/60b3c023fcc6401c8bfd6e7a2efe12d8 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"eb95bbc820e1507a05783d291705628487f4d54e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 13:59:47.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"89eb73aac86514c9293c3d83c4c45fdd88da625a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 14:11:48.000000000","message":"Topic changed from paunch/retire to paunch/retirement","accounts_in_message":[],"_revision_number":3},{"id":"713abf856690b9b89f31f6a38ac8628a94ae242a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 14:14:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d39c2d1e639f4ef5b1b9af8e999f5462 : SUCCESS in 8m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f967154bb7dd511d1232e4b0e34af7c3294ff0e1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 14:55:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7d5e5934e8ce627e91ea990c347656f57fa40e62","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-10 15:12:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0eab88925dad5073ee65b01ad0da5433506e3610","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 15:17:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dca5da067d9d52542d0351f653c83350c60145eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-10 15:22:05.000000000","message":"Patch Set 3: Code-Review+1\n\n\u003e (1 comment)\n\ngot it. syncing github for code removal also. +1 on removing it at last.","accounts_in_message":[],"_revision_number":3},{"id":"560d89fb252b61d5c5d4066c50c45678e1c8938f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 15:31:20.000000000","message":"Patch Set 3: Code-Review-1\n\nAs discussed on #openstack-infra, this is not a full retirement.\n\nInstead only master branch is retired. So, this needs the changes as documented in https://review.opendev.org/734877","accounts_in_message":[],"_revision_number":3},{"id":"40bffe68233b27280d120be363934d42aed461a1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 15:31:30.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"4cf78fd65891fefbf0e8e8ffd3b41054c0cab251","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 17:49:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9f12aaf2b9b565b8e478428b082de80d9105a88c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 17:55:39.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0bad20a983764c3b2153be8fb7ec93ca2b5ba1f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 18:03:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/71767f100ee54edfa029f4d42e3e180c : SUCCESS in 7m 07s","accounts_in_message":[],"_revision_number":4},{"id":"93cbec0e1895c7a29b6870d2d52c903538e63318","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-06-10 18:51:04.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8d3d9899291d8c385ce1b9411b02dd81c223ee7b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-12 14:59:02.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"3485997f78b3feec9545621b1e5530eeae921872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 14:59:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0bcb3deaa0ab41d6f45e2204de4a185c8e56574c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 15:07:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"3b64e1a258e62571c337206135d4ed88e36c4554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 15:07:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bfbcda98b2ee470a845f1237ecef5bc1 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5d6b3cacdfa605410b4a6e4076770bab31871560","revisions":{"82cc450955f60222b8a0ee82a91913c585ef5c84":{"kind":"REWORK","_number":1,"created":"2020-06-09 16:14:19.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/40/734640/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/734640/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/734640/1"}}},"commit":{"parents":[{"commit":"b59a59da354233af247b611f616894484616cd10","subject":"Use python 3 venv for generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b59a59da354233af247b611f616894484616cd10"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-09 14:30:19.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-09 14:30:19.000000000","tz":-240},"subject":"Retire Paunch","message":"Retire Paunch\n\nUse noop-jobs for Paunch, since it\u0027s now retired.\n\nChange-Id: Idaf5c54c101f3af0dcbf42e23caae236f5731a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82cc450955f60222b8a0ee82a91913c585ef5c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82cc450955f60222b8a0ee82a91913c585ef5c84"}]},"branch":"refs/heads/master"},"e5245a4814644a5ed41859ed9c6039d95bee064a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-10 13:27:25.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/40/734640/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/734640/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/734640/2"}}},"commit":{"parents":[{"commit":"b59a59da354233af247b611f616894484616cd10","subject":"Use python 3 venv for generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b59a59da354233af247b611f616894484616cd10"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-09 14:30:19.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 13:27:23.000000000","tz":0},"subject":"Retire Paunch","message":"Retire Paunch\n\nUse noop-jobs for Paunch, since it\u0027s now retired.\nThis was announced here:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/015111.html\n\nChange-Id: Idaf5c54c101f3af0dcbf42e23caae236f5731a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e5245a4814644a5ed41859ed9c6039d95bee064a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e5245a4814644a5ed41859ed9c6039d95bee064a"}]},"branch":"refs/heads/master"},"f0f40e881c25a8cb27e1b8106a2228bb8e15f85a":{"kind":"REWORK","_number":3,"created":"2020-06-10 13:59:47.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/40/734640/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/734640/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/734640/3"}}},"commit":{"parents":[{"commit":"b59a59da354233af247b611f616894484616cd10","subject":"Use python 3 venv for generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b59a59da354233af247b611f616894484616cd10"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-09 14:30:19.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 13:59:45.000000000","tz":0},"subject":"Retire Paunch","message":"Retire Paunch\n\nUse noop-jobs for Paunch, since it\u0027s now retired.\nThis was announced here:\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-May/015111.html\n\nChange-Id: Idaf5c54c101f3af0dcbf42e23caae236f5731a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0f40e881c25a8cb27e1b8106a2228bb8e15f85a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f0f40e881c25a8cb27e1b8106a2228bb8e15f85a"}]},"branch":"refs/heads/master"},"5d6b3cacdfa605410b4a6e4076770bab31871560":{"kind":"REWORK","_number":4,"created":"2020-06-10 17:49:50.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/40/734640/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/40/734640/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/40/734640/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/40/734640/4"}}},"commit":{"parents":[{"commit":"b59a59da354233af247b611f616894484616cd10","subject":"Use python 3 venv for generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b59a59da354233af247b611f616894484616cd10"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-09 14:30:19.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-06-10 17:49:25.000000000","tz":-240},"subject":"Deprecate Paunch","message":"Deprecate Paunch\n\nUse noop-jobs for Paunch, since it\u0027s not developped anymore.\n\nChange-Id: Idaf5c54c101f3af0dcbf42e23caae236f5731a6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d6b3cacdfa605410b4a6e4076770bab31871560"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5d6b3cacdfa605410b4a6e4076770bab31871560"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
