)]}'
{"id":"openstack%2Fproject-config~734791","triplet_id":"openstack%2Fproject-config~master~I0b04d92de6287b78bc422dd177403570eb18294a","project":"openstack/project-config","branch":"master","topic":"centos-stream","hashtags":[],"change_id":"I0b04d92de6287b78bc422dd177403570eb18294a","subject":"CentOS 8 Stream initial deployment","status":"MERGED","created":"2020-06-10 10:40:31.000000000","updated":"2020-10-06 11:13:51.000000000","submitted":"2020-10-06 06:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734791-1601965952694-3590af05","meta_rev_id":"b9c9fb28781143c756a56eb5716d0d538815316f","_number":734791,"virtual_id_number":734791,"owner":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-06 06:32:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-10-06 06:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-10-03 10:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2020-10-03 10:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-06 03:59:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-10-06 06:22:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-02 14:19:00.000000000","updated_by":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"reviewer":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"state":"REVIEWER"},{"updated":"2020-10-02 14:36:11.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-10-02 14:39:33.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2020-10-02 18:19:56.000000000","updated_by":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"reviewer":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"state":"REVIEWER"},{"updated":"2020-10-06 03:59:04.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-10-06 06:22:03.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-10-06 06:32:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3d7c7d09b255c52614c631aeb79b777d52930b4a","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-06-10 10:40:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22b3c55ff2c8e15a9d53bb88e657af8ee1aa4855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 10:51:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b6e79c7300994573b2f76f7de568794a : SUCCESS in 9m 56s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/a2d8be06a37346c9bd2a16ee0a535b00 : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"40df7cfa560a535893baf6aed35733039795577d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 17:40:58.000000000","message":"Patch Set 1: Code-Review-1\n\nwhy do we need another centos8 image? One should be enough...\n\nPlease update the commit message to explain why another is needed and the current image does not fulfill your needs and which are these needs.","accounts_in_message":[],"_revision_number":1},{"id":"72b3849fe4031e3c752cb98f1474f6194b4dff94","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-16 06:49:08.000000000","message":"Patch Set 1: Workflow-1\n\nstream is the rolling pre-release of centos-8, so useful as a heads-up on upcoming issues before point-releases break. this can be explained; however thiis is reliant on dib support being released too; https://review.opendev.org/734083","accounts_in_message":[],"_revision_number":1},{"id":"aada5328af45c7f25ecf5b6ed7108b005b31ba0f","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-07 09:59:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"db94a7ff498960a37f8dff43e8eff0bab463215a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:17:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/47148ee9a7fe40a6af4c42db3e13e49b : SUCCESS in 9m 37s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/972033e15d6d4cc19f1ea37a0b2deb88 : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"91b4eca066c13d2f3544ca52211b1df245a424e0","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-08-07 10:42:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"468eca2570b328503d50516801ab0739113b6aae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 10:53:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/00d591364db74fe797d129e4286d9e53 : SUCCESS in 7m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/5da31fb1a69b4e7c8fde3e311d9df001 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"02ebfc7c4a4f4363359580190d6a91698f49e96c","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-09-22 05:50:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"21d568074934507e7e46ff63e8dd07e4035b4bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-22 06:02:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eeb7d3e4d9744d7eb6ac4c40318523e0 : SUCCESS in 5m 46s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/fef929cbda7943c2aaa9f9e95fd8cf01 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"22c9fecc311eba7a65a4abd0b82ce5aae52e371d","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-10-02 14:18:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7af14cb01bbf0bb875d820fc012090c71f86f17c","author":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"date":"2020-10-02 14:19:00.000000000","message":"Patch Set 5:\n\nThe review to provide dib support was already merged, and this centos-8-stream image can be very useful to some deployment projects, like TripleO.","accounts_in_message":[],"_revision_number":5},{"id":"ad60337ce7bbe3cd8f90a042ca6a38b59911b4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 14:31:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/65cb0dcdb8024dbdb0e562911f397554 : SUCCESS in 5m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b45bc73f3a744c729037e491f13912ba : SUCCESS in 8m 08s","accounts_in_message":[],"_revision_number":5},{"id":"35c19f4480fbc72ead98bc00b22a723eee6e8190","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-10-02 14:36:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"66d220cbe1c12fb3ce3b26702bec308e6b35e8b5","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2020-10-02 14:39:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e388e96dd92bf8016c40163e8d99eb1a6588634c","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2020-10-02 18:10:15.000000000","message":"Patch Set 5:\n\nIt seems like openstack/project-config needs to be adapted for the new `release` value. This image build is failing locally with:\n\n  /tmp/in_target.d/post-install.d/80-enable-infra-services: line 20: stream: unbound variable","accounts_in_message":[],"_revision_number":5},{"id":"8d8080b856fe0b327c36ce33d8cb9b845655ac76","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2020-10-02 18:19:56.000000000","message":"Patch Set 5:\n\nIt seems like this needs to depends on https://review.opendev.org/755860","accounts_in_message":[],"_revision_number":5},{"id":"89f8594c054ba660bd0599e7abdb7574b26df914","author":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"date":"2020-10-03 10:38:34.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"46734b8c4c895a1021dd3007edd9e6ca8484a710","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-03 10:43:26.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b3d46fa03e0048f2907f144235e88380 : SUCCESS in 4m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b336ee9665284968acbca4668975bd74 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":6},{"id":"278003d61af60a002bccb263dadc9cb75eab4a3d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-10-06 03:59:04.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a31725347ca2bcbdde30d698d595af99a37b6e9f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-10-06 06:22:03.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f8376fa9306cb2b2e1886ce6a19014b55db5fff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:22:16.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9d777dd3f44d62d61acabd964ff5cb2fc12e5efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:32:32.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/310690d29f984c038d3727ecb053e9eb : SUCCESS in 8m 43s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/869f224cd33848d285b727866f6d2e7b : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":6},{"id":"7a483fc4b92779438326c953df892936850a5303","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 06:32:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"b87218b07caf7b19ebaed20ff2d8a68713a3203c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-06 11:13:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/c79763ec216d4f96ba49cd1990c264c8 : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"374b24a4bd0d35f22569ea46556c8a1c6d56f907","revisions":{"9e4d0702cd62bccc498b3081412519eddaeaa11e":{"kind":"REWORK","_number":1,"created":"2020-06-10 10:40:31.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/734791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/734791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/734791/1"}}},"commit":{"parents":[{"commit":"7fc1d3f88070301c0710910d7646cfab2e5b7481","subject":"Revert \"Wheel publish jobs: include system-config roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fc1d3f88070301c0710910d7646cfab2e5b7481"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"subject":"CentOS 8 Stream initial deployment","message":"CentOS 8 Stream initial deployment\n\nChange-Id: I0b04d92de6287b78bc422dd177403570eb18294a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e4d0702cd62bccc498b3081412519eddaeaa11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e4d0702cd62bccc498b3081412519eddaeaa11e"}]},"branch":"refs/heads/master"},"803abee30087c606d18a974f2a11266cc27610aa":{"kind":"REWORK","_number":2,"created":"2020-08-07 09:59:18.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/734791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/734791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/734791/2"}}},"commit":{"parents":[{"commit":"e4120f97907a4e1cc6d9e8e6f29dfd4b75b5cf44","subject":"revert https://review.opendev.org/739605","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e4120f97907a4e1cc6d9e8e6f29dfd4b75b5cf44"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-07 09:58:37.000000000","tz":120},"subject":"CentOS 8 Stream initial deployment","message":"CentOS 8 Stream initial deployment\n\nChange-Id: I0b04d92de6287b78bc422dd177403570eb18294a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/803abee30087c606d18a974f2a11266cc27610aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/803abee30087c606d18a974f2a11266cc27610aa"}]},"branch":"refs/heads/master"},"88ec77fdfe1abaa6b25a68e95b4fd154db8c1420":{"kind":"REWORK","_number":3,"created":"2020-08-07 10:42:44.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/734791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/734791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/734791/3"}}},"commit":{"parents":[{"commit":"e4120f97907a4e1cc6d9e8e6f29dfd4b75b5cf44","subject":"revert https://review.opendev.org/739605","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e4120f97907a4e1cc6d9e8e6f29dfd4b75b5cf44"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-08-07 10:42:33.000000000","tz":120},"subject":"CentOS 8 Stream initial deployment","message":"CentOS 8 Stream initial deployment\n\nChange-Id: I0b04d92de6287b78bc422dd177403570eb18294a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88ec77fdfe1abaa6b25a68e95b4fd154db8c1420"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88ec77fdfe1abaa6b25a68e95b4fd154db8c1420"}]},"branch":"refs/heads/master"},"fccdf35454491767339a53c153a50973d388799c":{"kind":"REWORK","_number":4,"created":"2020-09-22 05:50:17.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/734791/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/734791/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/734791/4"}}},"commit":{"parents":[{"commit":"ae4c1822c273039930e316774839aaf94d568303","subject":"Merge \"Switch grafana nodepool to Fedora 32\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae4c1822c273039930e316774839aaf94d568303"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-09-22 05:49:04.000000000","tz":120},"subject":"CentOS 8 Stream initial deployment","message":"CentOS 8 Stream initial deployment\n\nChange-Id: I0b04d92de6287b78bc422dd177403570eb18294a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fccdf35454491767339a53c153a50973d388799c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fccdf35454491767339a53c153a50973d388799c"}]},"branch":"refs/heads/master"},"d859ba9bb3bb13c424e21237d9e93791af03ea4f":{"kind":"REWORK","_number":5,"created":"2020-10-02 14:18:15.000000000","uploader":{"_account_id":13294,"name":"Javier Peña","email":"jpena@redhat.com","username":"jpena"},"ref":"refs/changes/91/734791/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/734791/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/734791/5"}}},"commit":{"parents":[{"commit":"c50d61d6fc208ca58781e88a9cf543bcb0473dbe","subject":"Merge \"Cleanup unneeded lines for os-loganalyze\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c50d61d6fc208ca58781e88a9cf543bcb0473dbe"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"committer":{"name":"Javier Pena","email":"jpena@redhat.com","date":"2020-10-02 14:18:13.000000000","tz":120},"subject":"CentOS 8 Stream initial deployment","message":"CentOS 8 Stream initial deployment\n\nChange-Id: I0b04d92de6287b78bc422dd177403570eb18294a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d859ba9bb3bb13c424e21237d9e93791af03ea4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d859ba9bb3bb13c424e21237d9e93791af03ea4f"}]},"branch":"refs/heads/master"},"374b24a4bd0d35f22569ea46556c8a1c6d56f907":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-10-03 10:38:34.000000000","uploader":{"_account_id":6469,"name":"Carlos Gonçalves","display_name":"Carlos Goncalves","email":"cgoncalves@redhat.com","username":"cgoncalves"},"ref":"refs/changes/91/734791/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/91/734791/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/91/734791/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/91/734791/6"}}},"commit":{"parents":[{"commit":"66cb2317fd04bf2562579751f40e12d542553b3b","subject":"infra-package-needs: add support for centos-8-stream","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/66cb2317fd04bf2562579751f40e12d542553b3b"}]}],"author":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-06-10 10:39:49.000000000","tz":120},"committer":{"name":"Carlos Goncalves","email":"cgoncalves@redhat.com","date":"2020-10-03 10:38:34.000000000","tz":0},"subject":"CentOS 8 Stream initial deployment","message":"CentOS 8 Stream initial deployment\n\nChange-Id: I0b04d92de6287b78bc422dd177403570eb18294a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/374b24a4bd0d35f22569ea46556c8a1c6d56f907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/374b24a4bd0d35f22569ea46556c8a1c6d56f907"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
