)]}'
{"id":"openstack%2Fproject-config~734863","triplet_id":"openstack%2Fproject-config~master~Iecb7534d742d186326f73171692515220af57596","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Iecb7534d742d186326f73171692515220af57596","subject":"Add openstack/magnum and openstack/magnum-tempest-plugin in required-projects","status":"MERGED","created":"2020-06-10 14:43:45.000000000","updated":"2020-06-11 04:57:43.000000000","submitted":"2020-06-11 04:47:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734863-1591850842013-ffb2baeb","meta_rev_id":"86045005abd93de9e64db6837a85a6122abdbff5","_number":734863,"virtual_id_number":734863,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 04:47:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-06-10 20:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-06-11 04:37:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-06-11 04:37:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 16:49:26.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-10 20:30:07.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-06-11 04:37:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-06-11 04:47:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2ab0e5f676e4f678fdbd913673e748c9d3d1e83d","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-06-10 14:43:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ce0ff477cae2ace1cef6e7245e545e880dcc225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 14:58:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4ddf86e3cc0f407e81433627abc0f62d : SUCCESS in 8m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b2111f45764c444f9b23bc8e37da1889 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8953b2b901f95f6f0274dfe2ac4cea42a67d8287","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-10 16:49:26.000000000","message":"Patch Set 1: Code-Review-1\n\nWhy?\n\nPLease explain in the commit message *why* you do this. That you do it is clear. But in the future, somebody might need to understand why this was done and go back to the commit message - and this one is just saying what without any reason.","accounts_in_message":[],"_revision_number":1},{"id":"32477f37bc29870b991744e2effb9785bf164b76","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-06-10 17:00:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3b7d89ca685d527c4cc5b355109a3742e1ab16f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:17:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d428ff4f9d3042dda80401a18ef0c926 : SUCCESS in 9m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ea02eefa19f74ae6b0e15c7c49831075 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"8f00d75d1ad11f808a23ff7b7305f81633e7cb91","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-06-10 20:30:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"63b42f6749eed0f87b1fff98cd11c00d04178148","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-11 04:37:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e41f09e7d49c401030f6098f075c45c38fe15451","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:38:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ada854484518e314b60a48a88705572f6270e28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:47:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d56b5f16ea024b7d9c47cafc55707c0a : SUCCESS in 8m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/018b9515369e4839937d60ca927a2496 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":2},{"id":"e2dad50dc736fd27a9ac4492ef815a41bf8d73fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:47:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"86045005abd93de9e64db6837a85a6122abdbff5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 04:57:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/648541ce42e849a89efc9c6fe623c1c0 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f69804f621db9ae30a5f17d2b54b6561565cc919","revisions":{"93e11e80d301f57b578d5ae152528b4bbbe6fee7":{"kind":"REWORK","_number":1,"created":"2020-06-10 14:43:45.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/63/734863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/734863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/734863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/734863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/734863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/734863/1"}}},"commit":{"parents":[{"commit":"7fc1d3f88070301c0710910d7646cfab2e5b7481","subject":"Revert \"Wheel publish jobs: include system-config roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fc1d3f88070301c0710910d7646cfab2e5b7481"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-06-10 14:42:36.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-06-10 14:43:41.000000000","tz":180},"subject":"Add magnum and magnum-tempest-plugin in required-projects","message":"Add magnum and magnum-tempest-plugin in required-projects\n\nChange-Id: Iecb7534d742d186326f73171692515220af57596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93e11e80d301f57b578d5ae152528b4bbbe6fee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93e11e80d301f57b578d5ae152528b4bbbe6fee7"}]},"branch":"refs/heads/master"},"f69804f621db9ae30a5f17d2b54b6561565cc919":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-10 17:00:32.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/63/734863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/63/734863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/734863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/734863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/63/734863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/63/734863/2"}}},"commit":{"parents":[{"commit":"7fc1d3f88070301c0710910d7646cfab2e5b7481","subject":"Revert \"Wheel publish jobs: include system-config roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7fc1d3f88070301c0710910d7646cfab2e5b7481"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-06-10 14:42:36.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-06-10 16:59:54.000000000","tz":180},"subject":"Add openstack/magnum and openstack/magnum-tempest-plugin in required-projects","message":"Add openstack/magnum and openstack/magnum-tempest-plugin in required-projects\n\nAppend them in required-jobs of vexxhost tenant\n\nChange-Id: Iecb7534d742d186326f73171692515220af57596\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f69804f621db9ae30a5f17d2b54b6561565cc919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f69804f621db9ae30a5f17d2b54b6561565cc919"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
