)]}'
{"id":"openstack%2Fproject-config~735301","triplet_id":"openstack%2Fproject-config~master~Ia14c2733272ea4c108dccef4b5c724769ed22a01","project":"openstack/project-config","branch":"master","topic":"centos-7-plain","hashtags":[],"change_id":"Ia14c2733272ea4c108dccef4b5c724769ed22a01","subject":"Add centos-7 plain nodes","status":"MERGED","created":"2020-06-12 07:20:11.000000000","updated":"2020-06-13 09:26:04.000000000","submitted":"2020-06-13 09:16:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735301-1592039760283-81340386","meta_rev_id":"62c62a9af6ba98bf5a34bde4b9e9198875c24176","_number":735301,"virtual_id_number":735301,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-13 09:16:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-06-13 05:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-13 09:09:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-13 05:03:26.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-13 09:09:13.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-06-13 09:16:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83b867fab8c41697a2a329523e5018d3c3458ee4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-12 07:20:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d381fae91ddfc841e7248b5c710d07badafd482d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 07:30:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4c396aee2dce43e186276dc4a4f229f6 : SUCCESS in 7m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/dcdc8158fda54f9a82a2c977eb4f0c36 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"fc30242c21596dda3ad3ac476d915a30dd0cee1a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-12 08:29:55.000000000","message":"Patch Set 1: Code-Review-1\n\nThat\u0027s not building anything, run a grep for centos-8-plain to see what else needs to be done.","accounts_in_message":[],"_revision_number":1},{"id":"46959381ff9b2a26b07b84059a66053d6bab9cdf","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-12 08:33:58.000000000","message":"Patch Set 1:\n\nI\u0027m not officially at work again until monday, which is when the -plain changes are scheduled for. If someone else who understands this could take a look it would be appreciated.","accounts_in_message":[],"_revision_number":1},{"id":"7670e28c0b6d944744d99afe4e0c6df58ebaf152","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-12 08:58:53.000000000","message":"Patch Set 1:\n\nWould others in osa team test with the image? Otherwise there\u0027s no sense in doing it","accounts_in_message":[],"_revision_number":1},{"id":"65026f16dd8776111e4a38815aa78b106c352cf7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-13 00:22:12.000000000","message":"Patch Set 1:\n\nsorry about this; i actually had a change from 20th march to fix this and must have got distracted and forgot to push it.  i\u0027ll upload it now.","accounts_in_message":[],"_revision_number":1},{"id":"d318a38ddafcce0b4ce43b8e4c6ac72084776ad6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-13 00:22:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2ff3564b7c8bf5cd54973733e21d9105871cd11d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 00:39:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/132c84aae9a840e6bc389ed9b21b36b7 : SUCCESS in 7m 23s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2073e6d0d6404e918f691493b1aab8bf : SUCCESS in 15m 52s","accounts_in_message":[],"_revision_number":2},{"id":"f07bc803079b9369eac7418304ea6384016e9a71","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-13 05:03:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"33710b0f1d4d655bb3956fc457763a79ccce3ff3","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-13 09:09:13.000000000","message":"Patch Set 2: Workflow+1\n\ni\u0027ll add this with ajaeger\u0027s overview and check it builds","accounts_in_message":[],"_revision_number":2},{"id":"79e17faa4b3aef23c3e13695f6e4c4cc67a3be89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 09:09:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bc8b20ada1abebef121c6520f3b0b01866dd2ebd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 09:16:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/80b4f1a6262c45e08774a7dcf08978e9 : SUCCESS in 6m 24s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/c1c262ee09864c548d3940dcf52ea932 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"3e65a92958048d569b35470c78909b01dc3211ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 09:16:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"89e07dd56c64c4e82b50ad52388eca1db986ce7d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 09:26:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/53b345db56bd41b1b505c1fa2488179f : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ec916fa40d93c008624fef250d6dbb6f7bfb1cf","revisions":{"f428f26e2e035cc3e33ee696b2d26333784d5f52":{"kind":"REWORK","_number":1,"created":"2020-06-12 07:20:11.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/01/735301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/735301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/735301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/735301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/735301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/735301/1"}}},"commit":{"parents":[{"commit":"6f2b723a9cf84baa0bfb272cd3133a098fed217b","subject":"Merge \"Install git-review under python3 for proposed updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f2b723a9cf84baa0bfb272cd3133a098fed217b"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-12 07:18:06.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-12 07:19:53.000000000","tz":60},"subject":"Add missing centos-7-plain disk images","message":"Add missing centos-7-plain disk images\n\nChange-Id: Ia14c2733272ea4c108dccef4b5c724769ed22a01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f428f26e2e035cc3e33ee696b2d26333784d5f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f428f26e2e035cc3e33ee696b2d26333784d5f52"}]},"branch":"refs/heads/master"},"8ec916fa40d93c008624fef250d6dbb6f7bfb1cf":{"kind":"REWORK","_number":2,"created":"2020-06-13 00:22:47.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/01/735301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/01/735301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/735301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/735301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/01/735301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/01/735301/2"}}},"commit":{"parents":[{"commit":"7f10d908e622fbd8ac12a3a224d7047cef341150","subject":"Retire pabelanger as infra-root","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7f10d908e622fbd8ac12a3a224d7047cef341150"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-05-26 03:15:28.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-13 00:22:23.000000000","tz":600},"subject":"Add centos-7 plain nodes","message":"Add centos-7 plain nodes\n\nI\u0027m not quite sure why I didn\u0027t add this with the other -plain nodes;\nadd it for completeness.\n\nChange-Id: Ia14c2733272ea4c108dccef4b5c724769ed22a01\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ec916fa40d93c008624fef250d6dbb6f7bfb1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ec916fa40d93c008624fef250d6dbb6f7bfb1cf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
