)]}'
{"id":"openstack%2Fproject-config~735426","triplet_id":"openstack%2Fproject-config~master~Iab246e98bf3fb187b54934c376a6ee3075f1a9c4","project":"openstack/project-config","branch":"master","topic":"interop-repo-renaming","hashtags":[],"change_id":"Iab246e98bf3fb187b54934c376a6ee3075f1a9c4","subject":"Fix wrong refstack gerrit ACL","status":"MERGED","created":"2020-06-13 14:12:32.000000000","updated":"2020-06-13 14:37:40.000000000","submitted":"2020-06-13 14:25:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735426-1592058329829-2895ad4b","meta_rev_id":"211663a5476f7d78ba70f23f29a1beafc47e726a","_number":735426,"virtual_id_number":735426,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-13 14:25:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-13 14:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-13 14:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-13 14:18:47.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-06-13 14:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c11f6c1d75ebfcf4d923055e50430fee80064dc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-13 14:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"979450f266d1339be87cd70ca0c0c9e27547de97","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-06-13 14:18:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9d88d4060f6c70d858436fd57e4fd8ec8d55cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 14:18:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/00cbe456bc0c4f70b4d7d668931f405b : SUCCESS in 5m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5b3082ca393043b688403a4fca03e029 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"55f47f0af0ee1f8bc6ecb743198d1a38b04444de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 14:18:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"134bc65cc1cfc1c0483a74491e5d94a69742d77e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-13 14:23:33.000000000","message":"Topic changed from interop-repo-rename to interop-repo-renaming","accounts_in_message":[],"_revision_number":1},{"id":"780aea183bcfd303c20563d7c806aab7194bc20c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 14:25:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"14a1a6c11775672298b2a3c648e72d9e2f4fe90a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 14:25:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/550dd3a5cac246be9739b27261c57b6f : SUCCESS in 5m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c1c66817d236498b86f42f57190aa8be : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"211663a5476f7d78ba70f23f29a1beafc47e726a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-13 14:37:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/858e23190305480f9cca93a646c11498 : SUCCESS in 2m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ebbb1e9fc0b3888cb9c3fff9638bfa4cd8613e3","revisions":{"7ebbb1e9fc0b3888cb9c3fff9638bfa4cd8613e3":{"kind":"REWORK","_number":1,"created":"2020-06-13 14:12:32.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/26/735426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/26/735426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/735426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/735426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/26/735426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/26/735426/1"}}},"commit":{"parents":[{"commit":"212a955f8882221cdbccbe98a633adcde5a5deac","subject":"Merge \"Add centos-7 plain nodes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/212a955f8882221cdbccbe98a633adcde5a5deac"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-13 14:11:31.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-06-13 14:11:31.000000000","tz":120},"subject":"Fix wrong refstack gerrit ACL","message":"Fix wrong refstack gerrit ACL\n\nChange I02ac67e333cc1a5133269bd8562f18a8d50bb036 moved\ngerrit/acls/openstack/refstack.yaml to gerrit/acls/osf/refstack.yaml,\nupdate a missing reference.\n\nChange-Id: Iab246e98bf3fb187b54934c376a6ee3075f1a9c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ebbb1e9fc0b3888cb9c3fff9638bfa4cd8613e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7ebbb1e9fc0b3888cb9c3fff9638bfa4cd8613e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
