)]}'
{"id":"openstack%2Fproject-config~735812","triplet_id":"openstack%2Fproject-config~master~Ib1a38e3fce44591cea41d04d3a15cdaae97124b4","project":"openstack/project-config","branch":"master","topic":"retire-neutron-fwaas","hashtags":[],"change_id":"Ib1a38e3fce44591cea41d04d3a15cdaae97124b4","subject":"Retire neutron-fwaas and neutron-fwaas-dashboard projects","status":"ABANDONED","created":"2020-06-16 08:53:48.000000000","updated":"2020-06-16 09:53:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ed16fd353c9b3aa65a6df29e0555ff77a392eb19","_number":735812,"virtual_id_number":735812,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 08:56:56.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-16 09:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d4610be0c5ae9e80b24b311b8ec188b751ac34f1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 08:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"daf4d2f4328e00f2e304f9ddfcec6846bd179c06","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-16 08:56:56.000000000","message":"Patch Set 1: Code-Review-1\n\nWe\u0027re not renaming to x/ namespace anymore for OpenStack projects, there was a TC decision on it.\n\nYou need to retire and then import again","accounts_in_message":[],"_revision_number":1},{"id":"a72d35d2d02f2896bf72b4e2266d8fdf0e2d1c1f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-16 08:58:49.000000000","message":"Patch Set 1:\n\nIf there are no maintainers, there\u0027s no need to move to x. Just retire it.","accounts_in_message":[],"_revision_number":1},{"id":"251b4cabef9f5e6aae001781a0645b199c055f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:12:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d366273ce88645cdab0bac8057a687c6 : SUCCESS in 8m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d6a04a2a1c8042159ba3e0f9334c0b9d : FAILURE in 7m 30s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c84a0d17ea6b47cb94d4af730dcf3ac8 : SUCCESS in 10m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b294afeca31a8444801d2e6afc7ca558009f8b53","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 09:47:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7df01d5d521cb6894283ca4ff0066aacaa21ba40","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-06-16 09:53:29.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c96467668913dfefd145bec2421e6df456525878","revisions":{"c1e92cc97a7cc1c688166a9050419d0dd13a4587":{"kind":"REWORK","_number":1,"created":"2020-06-16 08:53:48.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/12/735812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/735812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/735812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/735812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/735812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/735812/1"}}},"commit":{"parents":[{"commit":"515cd465fd756c8d1f074391f3c05924cd60b63f","subject":"Retire Tricircle projects: finish infra todo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/515cd465fd756c8d1f074391f3c05924cd60b63f"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 08:51:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 08:51:24.000000000","tz":120},"subject":"Rename neutron-fwaas and neutron-fwaas-dashboard to x/ namespace","message":"Rename neutron-fwaas and neutron-fwaas-dashboard to x/ namespace\n\nProject neutron-fwaas and its dashboard were in neutron stadium but due\nto lack of maintainers we decided to deprecate it in Ussuri.\nNow it\u0027s time to remove this project from neutron stadium and\nmove it to the x/ namespace.\n\nChange-Id: Ib1a38e3fce44591cea41d04d3a15cdaae97124b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1e92cc97a7cc1c688166a9050419d0dd13a4587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1e92cc97a7cc1c688166a9050419d0dd13a4587"}]},"branch":"refs/heads/master"},"c96467668913dfefd145bec2421e6df456525878":{"kind":"REWORK","_number":2,"created":"2020-06-16 09:47:24.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/12/735812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/735812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/735812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/735812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/735812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/735812/2"}}},"commit":{"parents":[{"commit":"d826f61f20bfedadefdfa3c325630dc363821d4e","subject":"Add noop-jobs for neutron-fwaas and neutron-fwaas-dashboard projects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d826f61f20bfedadefdfa3c325630dc363821d4e"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 08:51:24.000000000","tz":120},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2020-06-16 09:15:34.000000000","tz":120},"subject":"Retire neutron-fwaas and neutron-fwaas-dashboard projects","message":"Retire neutron-fwaas and neutron-fwaas-dashboard projects\n\nProject neutron-fwaas and its dashboard were in neutron stadium but due\nto lack of maintainers we decided to deprecate it in Ussuri.\nNow it\u0027s time to remove this project from neutron stadium and\nmove it to the x/ namespace.\n\nChange-Id: Ib1a38e3fce44591cea41d04d3a15cdaae97124b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c96467668913dfefd145bec2421e6df456525878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c96467668913dfefd145bec2421e6df456525878"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
