)]}'
{"id":"openstack%2Fproject-config~737133","triplet_id":"openstack%2Fproject-config~master~I048f7b736d44abd3b76f589d69048e2d2f7a78ce","project":"openstack/project-config","branch":"master","topic":"retire-congress","hashtags":[],"change_id":"I048f7b736d44abd3b76f589d69048e2d2f7a78ce","subject":"Retire the puppet-congress project - Step 1 End Project Gating","status":"MERGED","created":"2020-06-20 15:02:07.000000000","updated":"2020-06-22 13:26:00.000000000","submitted":"2020-06-22 13:26:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737133-1592832360411-0c805858","meta_rev_id":"396942deb659bd5fc2f6f76818d9a9b4838aa63d","_number":737133,"virtual_id_number":737133,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-20 18:07:49.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 13:26:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"recommended":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-06-22 12:58:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-22 13:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2020-06-22 13:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2020-06-22 13:14:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-20 18:07:49.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-22 12:58:18.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-06-22 13:10:22.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-06-22 13:14:37.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-06-22 13:26:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cfd2bc7b483f5f7d504f7599cd4c07f6ae4b7483","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-20 15:02:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d281cf6adc45167d741c2dbe82c4306845f9525","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-20 15:06:55.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"35932c33042ce2461b341ca8157300c0d5a366e5","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 15:07:03.000000000","message":"Patch Set 1:\n\nwas there an announcement about this retirement on openstack-discuss?\n\nPlease add link to it to commit message.","accounts_in_message":[],"_revision_number":1},{"id":"163a86d4a03fe57bbbddea99cd730b82ef3decc7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 15:08:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18109011eac2fa3cfd87a3b83029d4af4deb74c4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 15:08:38.000000000","message":"Patch Set 2: Code-Review-1\n\nmy comment on patch set 1 is still valid","accounts_in_message":[],"_revision_number":2},{"id":"151a214cae08fd34ca5c4ea4fd2619314d236d3e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 15:12:59.000000000","message":"Patch Set 2:\n\nYou need 4 changes for retirement: 2 for project-config, one for governance and one for your repo. Please push all of them so that we can review all together.","accounts_in_message":[],"_revision_number":2},{"id":"69f05b8ca8b07d11ec677ce6d5e695927ae83d3b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-20 15:14:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fb3d9997b404864ac5b89cf3f78bfccc36def750","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-20 15:14:15.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"8477e0ddfd20c1b6157e3367bbbc06f784bf5002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 15:20:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/025e213b6ca144e68d00cfc8aba1960f : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"666f0b5e7e6f794f6b451b40716af473f751cab1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-20 15:21:06.000000000","message":"Patch Set 3:\n\n@Andreas\n\nI don\u0027t think we have announcement sent independently for puppet-congress.\nI initially thought that retirement of congress would be enough reason to retire puppet-congress because it is clearly meaningless to maintain puppet modules for retired software.\nHowever I\u0027ll send that announcement if that is required from governance perspective.","accounts_in_message":[],"_revision_number":3},{"id":"221a61b45caed5219658c673ff9e0515b1c3a121","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 16:14:46.000000000","message":"Patch Set 3:\n\n\u003e @Andreas\n \u003e \n \u003e I don\u0027t think we have announcement sent independently for\n \u003e puppet-congress.\n \u003e I initially thought that retirement of congress would be enough\n \u003e reason to retire puppet-congress because it is clearly meaningless\n \u003e to maintain puppet modules for retired software.\n \u003e However I\u0027ll send that announcement if that is required from\n \u003e governance perspective.\n\nthe congress retirement is enough reason - still, announce it so that everybody is aware of it. It\u0027s not reason that I want, it\u0027s communication ;)","accounts_in_message":[],"_revision_number":3},{"id":"0825993573b0365b8592d1e27d95245980dd6cf6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 16:15:22.000000000","message":"Patch Set 3:\n\nWaiting for announcement and the two remaining changes.","accounts_in_message":[],"_revision_number":3},{"id":"303382a88b67201f241bebe76ab8ebc2310f7aef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-20 16:42:12.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"b9e2d4012f039888701b64b3fc1b424842f48d42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 16:52:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3ecf8ddfb011448e9935fd681d98121e : SUCCESS in 7m 04s","accounts_in_message":[],"_revision_number":4},{"id":"df8c67e8ec11836797e6980e9bafb44c74c76d11","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-20 18:07:49.000000000","message":"Patch Set 4:\n\nLGTM, waiting for PTL+1.","accounts_in_message":[],"_revision_number":4},{"id":"8773a667f6265925e27655a031415c664eed3640","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-06-22 12:58:18.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8908254b8927ff15be5b424df6c85bb574752e4c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-22 13:10:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"73680014bb89a42a3bdff5cebe4c44510c5f5d6a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-06-22 13:14:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5afde5ffd563c045a4d79f14ed3e15fc9f531f6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 13:14:54.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e3f66532dc5335a3db468afdb68fd497473ad24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 13:26:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e17f2d124f294d969763970dd5e63dbc : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":4},{"id":"00d733daae9472eb2f5b207b2cad96492868bf9d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 13:26:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"481dac69996f65ab684bfc00519a60f341e2264b","revisions":{"e01aa9a59c5b48f11935c53f9ed518af5d60ce82":{"kind":"REWORK","_number":1,"created":"2020-06-20 15:02:07.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/737133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/737133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/737133/1"}}},"commit":{"parents":[{"commit":"c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0","subject":"Reduce max servers on OE - air handler needs repairs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:01:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:02:05.000000000","tz":540},"subject":"Add no-op jobs for puppet-congress preparatory to retirement","message":"Add no-op jobs for puppet-congress preparatory to retirement\n\nChange-Id: I048f7b736d44abd3b76f589d69048e2d2f7a78ce\nNeeded-by: https://review.opendev.org/#/c/731770/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e01aa9a59c5b48f11935c53f9ed518af5d60ce82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e01aa9a59c5b48f11935c53f9ed518af5d60ce82"}]},"branch":"refs/heads/master"},"4aedaeeb3d2b445716a62380fcb0965e19ef0dbf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-20 15:06:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/737133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/737133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/737133/2"}}},"commit":{"parents":[{"commit":"c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0","subject":"Reduce max servers on OE - air handler needs repairs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:01:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:06:52.000000000","tz":0},"subject":"Retire the puppet-congress project - Step 1 End Project Gating","message":"Retire the puppet-congress project - Step 1 End Project Gating\n\nUnfortunately Congress project has been retired already[1], so it is\nuseless to maintain puppet modules for it.\n\nChange-Id: I048f7b736d44abd3b76f589d69048e2d2f7a78ce\nNeeded-by: https://review.opendev.org/#/c/731770/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4aedaeeb3d2b445716a62380fcb0965e19ef0dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4aedaeeb3d2b445716a62380fcb0965e19ef0dbf"}]},"branch":"refs/heads/master"},"94ec4a6636a6fb6a7cb1abe2084f4e23369423f1":{"kind":"REWORK","_number":3,"created":"2020-06-20 15:14:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/737133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/737133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/737133/3"}}},"commit":{"parents":[{"commit":"c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0","subject":"Reduce max servers on OE - air handler needs repairs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:01:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:14:08.000000000","tz":0},"subject":"Retire the puppet-congress project - Step 1 End Project Gating","message":"Retire the puppet-congress project - Step 1 End Project Gating\n\nUnfortunately Congress project has been retired already[1], so it is\nuseless to maintain puppet modules for it.\n\nChange-Id: I048f7b736d44abd3b76f589d69048e2d2f7a78ce\nNeeded-by: https://review.opendev.org/#/c/731770/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94ec4a6636a6fb6a7cb1abe2084f4e23369423f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/94ec4a6636a6fb6a7cb1abe2084f4e23369423f1"}]},"branch":"refs/heads/master"},"481dac69996f65ab684bfc00519a60f341e2264b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-20 16:42:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/737133/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/33/737133/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/33/737133/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/33/737133/4"}}},"commit":{"parents":[{"commit":"c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0","subject":"Reduce max servers on OE - air handler needs repairs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5b40b1acb7f43b9b7f3258e08aa91a2c1f809e0"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 15:01:22.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-20 16:37:28.000000000","tz":540},"subject":"Retire the puppet-congress project - Step 1 End Project Gating","message":"Retire the puppet-congress project - Step 1 End Project Gating\n\nUnfortunately Congress project has been retired already[1], so it is\nuseless to maintain puppet modules for it.\n\nThe annoucement can be found in the following link.\n http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015555.html\n\n[1] https://review.opendev.org/#/c/721742/\n\nChange-Id: I048f7b736d44abd3b76f589d69048e2d2f7a78ce\nNeeded-by: https://review.opendev.org/#/c/731770/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/481dac69996f65ab684bfc00519a60f341e2264b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/481dac69996f65ab684bfc00519a60f341e2264b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
