)]}'
{"id":"openstack%2Fproject-config~737457","triplet_id":"openstack%2Fproject-config~master~Iafdc2f77a627969a39370833f084cd7555da96fd","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Iafdc2f77a627969a39370833f084cd7555da96fd","subject":"Upadting horizon nodejs job name","status":"MERGED","created":"2020-06-23 08:37:35.000000000","updated":"2020-07-02 05:53:24.000000000","submitted":"2020-07-02 05:53:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737457-1593669204708-b4231495","meta_rev_id":"28cdb223b3afbe92a0221d5cb62e727263e07ee7","_number":737457,"virtual_id_number":737457,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-06-24 12:12:53.000000000","_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 05:53:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":2,"date":"2020-06-25 18:51:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-02 05:44:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-02 05:44:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 12:12:53.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-06-25 18:51:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-02 05:44:23.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-07-02 05:53:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9861dd45781cc5f497a10891ff6e60564458780d","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 08:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92a08a272038fbc24e5c585d20794e27d7a38e98","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-23 08:44:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8a98fd3db6e95398e3ef0aef91336c8d9f7a1c66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 08:47:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/766ef31d405f4467a953d263353bfbbc : SUCCESS in 6m 19s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/2a027c7763cd4934ae1e376f145c2fc3 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d814ae267ffa319e5b57e534dc8dc1f8395fa418","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 08:52:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"89db6ead0062c558e7dfd373f7a03c975fccc8df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 09:00:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/45be27614e984bf7b144e0674f53cbeb : SUCCESS in 8m 06s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/082a3fdb581746ea904060c4534c3ec2 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2879293bd6f5b5508ea782880846777895ef0aac","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-23 11:21:23.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"468686863e56217fceac5c37e126ea063a62db0b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-24 06:11:48.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fb21c2ac65db7201b9d08565ea7f45cf8a37cb68","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-24 08:02:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"92be917e5e4acbdaa7098711f316024db48795f2","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-24 08:12:01.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c92abf96a666058d83023c97064aa7eb0d635e33","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-24 08:12:55.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"c225bcdbfcf3a30f43e882ef12c581e639606051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 08:26:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f24bfacc07444304b57b6c9ec816300d : SUCCESS in 8m 14s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/6353c11a75b947fdb98aeac204c6fbee : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"b30e188907f37d8718bad777deca3d195d0e86f1","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-24 11:32:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nI wonder why you didn\u0027t check your proposal matches the current horizon gate result....","accounts_in_message":[],"_revision_number":4},{"id":"41a44f2f2a44d9e1d5d34e26126d3494d866480b","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-24 11:33:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"745f94105c7268ae52a24b4b325974ba0880cefe","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-24 11:43:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"03688324b7f2e06fc254f3bc0ef4a0bd84e42657","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2020-06-24 11:44:34.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"e37b9f5e47d2e3438a67ad13270df5970e8ed905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:02:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ff32193918fc44ebabb455136bf3d92d : SUCCESS in 7m 34s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/43741a22431343d2873bd766ff923f79 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":6},{"id":"86d87724947257bd037453078e900222f9f6e08d","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-06-24 12:12:53.000000000","message":"Patch Set 6:\n\nI have no time to check one by one at graphite.o.o. this week, so I will review it next week (or this weekend).","accounts_in_message":[],"_revision_number":6},{"id":"d345ccaf8f30eee7f5740d0a985cbeaad4482ad1","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 18:51:38.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ad1ecc6f35ad198a31de01f97a179c8411f58ee9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-02 05:44:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5fb970cd9a91693e43f8150ecc1b056bbfff3c72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:44:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"369705c6b6e55ccd7c36bab92e4af69af74cbabd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:53:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"28cdb223b3afbe92a0221d5cb62e727263e07ee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:53:24.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f2b2323c952543868b832a1a48f4edce : SUCCESS in 7m 54s\n- project-config-grafyaml https://zuul.opendev.org/t/openstack/build/a0d1d03701754624aafcc31a6383464f : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b55c6d6737d65a2166eb5041f51b8a49d4774c82","revisions":{"fe8ea01f517e60948098b867a49ceca9a69ed00e":{"kind":"REWORK","_number":1,"created":"2020-06-23 08:37:35.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/737457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/737457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/737457/1"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:35:43.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:37:26.000000000","tz":0},"subject":"Upadting horizon nodejs job name","message":"Upadting horizon nodejs job name\n\nChange I86a56afcd2cde3a088abcc66c0294424cd03546f updates horizon\nnodejs job name and this patch to the same in horizon.yaml\nso that nodejs job data is visible on grafana dashboard.\n\nChange-Id: Iafdc2f77a627969a39370833f084cd7555da96fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe8ea01f517e60948098b867a49ceca9a69ed00e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe8ea01f517e60948098b867a49ceca9a69ed00e"}]},"branch":"refs/heads/master"},"e865f92ac0daaa140a441dc3ad6f26f8dd72614e":{"kind":"REWORK","_number":2,"created":"2020-06-23 08:52:12.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/737457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/737457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/737457/2"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:35:43.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:51:32.000000000","tz":0},"subject":"Upadting horizon nodejs job name","message":"Upadting horizon nodejs job name\n\nChange I86a56afcd2cde3a088abcc66c0294424cd03546f updates horizon\nnodejs job name and this patch to the same in horizon.yaml\nso that nodejs job data is visible on grafana dashboard.\n\nChange-Id: Iafdc2f77a627969a39370833f084cd7555da96fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e865f92ac0daaa140a441dc3ad6f26f8dd72614e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e865f92ac0daaa140a441dc3ad6f26f8dd72614e"}]},"branch":"refs/heads/master"},"f35b8031b754a6e2cf803d02e73a3dbea4b08685":{"kind":"REWORK","_number":3,"created":"2020-06-24 08:02:59.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/737457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/737457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/737457/3"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:35:43.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-24 07:48:10.000000000","tz":0},"subject":"Upadting horizon nodejs job name","message":"Upadting horizon nodejs job name\n\nChange I86a56afcd2cde3a088abcc66c0294424cd03546f updates horizon\nnodejs job name and this patch to the same in horizon.yaml\nso that nodejs job data is visible on grafana dashboard.\n\nThis patch also done a few cleaups:\n- Remove py27 jobs.\n- Remove python3-django111 jobs.\n- Added py38 and removed py37.\n\nChange-Id: Iafdc2f77a627969a39370833f084cd7555da96fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f35b8031b754a6e2cf803d02e73a3dbea4b08685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f35b8031b754a6e2cf803d02e73a3dbea4b08685"}]},"branch":"refs/heads/master"},"c31aa43f3f01692bec56572419e472506245a5b6":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-06-24 08:12:55.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/737457/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/737457/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/737457/4"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:35:43.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-06-24 08:12:45.000000000","tz":0},"subject":"Upadting horizon nodejs job name","message":"Upadting horizon nodejs job name\n\nChange I86a56afcd2cde3a088abcc66c0294424cd03546f updates horizon\nnodejs job name and this patch to the same in horizon.yaml\nso that nodejs job data is visible on grafana dashboard.\n\nThis patch also done a few cleanups:\n- Remove py27 jobs.\n- Remove python3-django111 jobs.\n- Added py38 and removed py37.\n\nChange-Id: Iafdc2f77a627969a39370833f084cd7555da96fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c31aa43f3f01692bec56572419e472506245a5b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c31aa43f3f01692bec56572419e472506245a5b6"}]},"branch":"refs/heads/master"},"1de9eb85bae9d12bd0a9abddfc09b670ef2cd024":{"kind":"REWORK","_number":5,"created":"2020-06-24 11:43:49.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/737457/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/737457/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/737457/5"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:35:43.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-24 11:31:24.000000000","tz":0},"subject":"Upadting horizon nodejs job name","message":"Upadting horizon nodejs job name\n\nChange I86a56afcd2cde3a088abcc66c0294424cd03546f updates horizon\nnodejs job name and this patch to the same in horizon.yaml\nso that nodejs job data is visible on grafana dashboard.\n\nThis patch also done a few cleaups:\n- Remove py27 jobs.\n- Remove python3-django111 jobs.\n- Remove python python3-django22 jobs.\n- Added py38 and removed py37.\n\nChange-Id: Iafdc2f77a627969a39370833f084cd7555da96fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1de9eb85bae9d12bd0a9abddfc09b670ef2cd024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1de9eb85bae9d12bd0a9abddfc09b670ef2cd024"}]},"branch":"refs/heads/master"},"b55c6d6737d65a2166eb5041f51b8a49d4774c82":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-06-24 11:44:34.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/57/737457/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/57/737457/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/57/737457/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/57/737457/6"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2020-06-23 08:35:43.000000000","tz":0},"committer":{"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","date":"2020-06-24 11:44:30.000000000","tz":0},"subject":"Upadting horizon nodejs job name","message":"Upadting horizon nodejs job name\n\nChange I86a56afcd2cde3a088abcc66c0294424cd03546f updates horizon\nnodejs job name and this patch to the same in horizon.yaml\nso that nodejs job data is visible on grafana dashboard.\n\nThis patch also done a few cleaups:\n- Remove py27 jobs.\n- Remove python3-django111 jobs.\n- Remove python3-django22 jobs.\n- Added py38 and removed py37.\n\nChange-Id: Iafdc2f77a627969a39370833f084cd7555da96fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b55c6d6737d65a2166eb5041f51b8a49d4774c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b55c6d6737d65a2166eb5041f51b8a49d4774c82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
