)]}'
{"id":"openstack%2Fproject-config~737636","triplet_id":"openstack%2Fproject-config~master~Iad74534d37bf88f64e6c04f4d34f3fba5504478f","project":"openstack/project-config","branch":"master","topic":"cleanup-project-retirements","hashtags":[],"change_id":"Iad74534d37bf88f64e6c04f4d34f3fba5504478f","subject":"Temporarily unretire incorrectly retired projects","status":"ABANDONED","created":"2020-06-23 21:29:32.000000000","updated":"2020-06-24 14:44:41.000000000","total_comment_count":79,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"939f0fabd31cf8f65e1f8faad138a269b32a50b7","_number":737636,"virtual_id_number":737636,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-23 21:44:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":1,"date":"2020-06-24 13:59:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-23 21:44:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-24 13:59:55.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"5f366301430a609d2225a8d41c6c22d46e1168cd","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-23 21:29:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a68fb083b4dedbb569940960be16444f92d5909","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:29:53.000000000","message":"Patch Set 1: Verified-1\n\n(79 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  The project \"opendev/puppet-apps_site\" was not found.  All projects\n  referenced within a Zuul configuration must first be added to the main\n  configuration file by the Zuul administrator.\n\nThe error appears in the following project stanza:\n\n  project:\n      name: opendev/puppet-apps_site\n      templates:\n        - noop-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 359, column 3","accounts_in_message":[],"_revision_number":1},{"id":"34c5ad225a58189194e70ecb153380120fa19a48","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-23 21:30:00.000000000","message":"Patch Set 1:\n\nthe commands for infra to run: http://paste.openstack.org/show/795110/","accounts_in_message":[],"_revision_number":1},{"id":"59649a533fd0de68c78c7723811d63d9563f2893","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-23 21:31:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f139ac04865505e2193b87c625b8cb66ffebbe06","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-23 21:35:22.000000000","message":"Topic changed from fix-ovh-keystone to cleanup-project-retirements","accounts_in_message":[],"_revision_number":2},{"id":"712cb4ad2577b9a383efc9522d9808af590f67dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 21:44:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4624e6d845b844c384d6f1937e800023 : SUCCESS in 8m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/58eebf20c4a44c6eb37856bb8a8c0b7b : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3c44a3e59deca1da704c4542f09324e9900e11aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-24 13:59:55.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM (should be rebased), do you think it will have an impact on release side [1]?\n\nLast week we ran this check [1] and everything was clean but I\u0027m not sure these changes (yours) will introduce some side effects on our releases team side.\n\n[1] https://releases.openstack.org/reference/using.html#tools-aclmanager-py","accounts_in_message":[],"_revision_number":2},{"id":"939f0fabd31cf8f65e1f8faad138a269b32a50b7","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-06-24 14:44:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5db424da2480e8495002b2d51f29bda9981bafe9","revisions":{"4e50fd59195ca99ab0d84c971cca5030d23db298":{"kind":"REWORK","_number":1,"created":"2020-06-23 21:29:32.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/36/737636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/737636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/737636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/737636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/737636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/737636/1"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-23 21:27:39.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-23 21:27:39.000000000","tz":-240},"subject":"Temporarily unretire incorrectly retired projects","message":"Temporarily unretire incorrectly retired projects\n\nWe have a few projects which have not been properly retired, this patch\nadds noop-jobs to all of those projects as well as changes their ACLs to\nallow TC members to land changes on those repos to help with the\nretirement and cleanup.\n\nThere will also be a list of commands provided to the OpenDev team to\nmanually make those projects read-write as they are currently in\nread-only state to allow pushing the proper retirement patches.\n\nChange-Id: Iad74534d37bf88f64e6c04f4d34f3fba5504478f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e50fd59195ca99ab0d84c971cca5030d23db298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4e50fd59195ca99ab0d84c971cca5030d23db298"}]},"branch":"refs/heads/master"},"5db424da2480e8495002b2d51f29bda9981bafe9":{"kind":"REWORK","_number":2,"created":"2020-06-23 21:31:52.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/36/737636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/36/737636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/737636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/737636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/36/737636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/36/737636/2"}}},"commit":{"parents":[{"commit":"83aa828d931724df9b03bd94222f820f3ada9b10","subject":"Merge \"Add os_adjutant zuul project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/83aa828d931724df9b03bd94222f820f3ada9b10"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-23 21:27:39.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-06-23 21:31:46.000000000","tz":-240},"subject":"Temporarily unretire incorrectly retired projects","message":"Temporarily unretire incorrectly retired projects\n\nWe have a few projects which have not been properly retired, this patch\nchanges their ACLs to allow TC members to land changes on those repos to\nhelp with the retirement and cleanup.\n\nThere will also be a list of commands provided to the OpenDev team to\nmanually make those projects read-write as they are currently in\nread-only state to allow pushing the proper retirement patches.\n\nChange-Id: Iad74534d37bf88f64e6c04f4d34f3fba5504478f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5db424da2480e8495002b2d51f29bda9981bafe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5db424da2480e8495002b2d51f29bda9981bafe9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
