)]}'
{"id":"openstack%2Fproject-config~737678","triplet_id":"openstack%2Fproject-config~master~I63df401c739022855e8e2f89e2dbc1b248eb792e","project":"openstack/project-config","branch":"master","topic":"release-loop","hashtags":[],"change_id":"I63df401c739022855e8e2f89e2dbc1b248eb792e","subject":"wheel-cache: convert release to a loop","status":"ABANDONED","created":"2020-06-24 04:59:21.000000000","updated":"2020-06-24 06:00:21.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3922e64029876bbf3f9bf8a2f52775c23892cab6","_number":737678,"virtual_id_number":737678,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-06-24 05:18:26.000000000","_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-06-24 05:07:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-06-24 04:59:57.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-06-24 05:46:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 05:07:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-24 05:18:26.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-24 05:46:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"186a95d0c8211340c16a02ce3048e8a35d6c941d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-24 04:59:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47314fe194162767e505a29bae1187c479bdf41a","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-24 04:59:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40717b152b37a2c94aaca333806b0fd807e88556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 05:07:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3eddd83e61894a179d18a541cc854cd0 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"eac9d169b1f828c5e5e8cec786843608540ab66d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-24 05:18:26.000000000","message":"Patch Set 1:\n\nI thought this was fixed by https://review.opendev.org/737525","accounts_in_message":[],"_revision_number":1},{"id":"3a110c0d57dfc9ad894bb6646f080e7d644dc7b1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-06-24 05:46:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nSorry for already having fixed the bug you were chasing, but this looks still like a good improvement.","accounts_in_message":[],"_revision_number":1},{"id":"867cf6ce0f071a820e4ead2975bedab9c256015e","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-06-24 06:00:21.000000000","message":"Abandoned\n\ni didn\u0027t check the queue, so the reason this already looks right is because it is.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53ef9d865c618451d6a660ecdb58413378a49fa2","revisions":{"53ef9d865c618451d6a660ecdb58413378a49fa2":{"kind":"REWORK","_number":1,"created":"2020-06-24 04:59:21.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/78/737678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/737678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/737678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/737678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/737678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/737678/1"}}},"commit":{"parents":[{"commit":"904576625292eae3bd51f2cc21e7262e0170a2e9","subject":"Merge \"grafana: don\u0027t use bool for refresh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/904576625292eae3bd51f2cc21e7262e0170a2e9"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-24 04:55:22.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-06-24 04:58:54.000000000","tz":600},"subject":"wheel-cache: convert release to a loop","message":"wheel-cache: convert release to a loop\n\nFor some reason I can not understand, we see\n\n  The task includes an option with an undefined variable. The error\n  was: \u0027afs_volume\u0027 is undefined\n\nwhen this gets up to \u0027mirror.wheel.busterx64\u0027 (i.e. the first centos\nreleases happen fine!).\n\nI can not see anything different about this (typo, etc.).  I can not\nreplicate this.  Unfortunately this has been broken for a while, and\nwe don\u0027t appear to have any logs from the last time it ran\nsuccessfully.  However, I think we have upgraded to Ansible 2.9 some\ntime in between and possibly this is related.\n\nThe format used here for the roles is old anyway, so convert it to\ninclude_role and put it in a loop -- hopefully this just magically\ndisappears; or otherwise we\u0027ll have something else to debug.\n\nChange-Id: I63df401c739022855e8e2f89e2dbc1b248eb792e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53ef9d865c618451d6a660ecdb58413378a49fa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53ef9d865c618451d6a660ecdb58413378a49fa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
