)]}'
{"id":"openstack%2Fproject-config~737941","triplet_id":"openstack%2Fproject-config~master~Ifec9058bda07c9f50fb3d7178067142828f4ce33","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ifec9058bda07c9f50fb3d7178067142828f4ce33","subject":"Slowly Scale OE back up","status":"MERGED","created":"2020-06-25 09:59:41.000000000","updated":"2020-06-25 10:38:05.000000000","submitted":"2020-06-25 10:27:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737941-1593080876416-5d897941","meta_rev_id":"042fc62204532905fc4819942b60bff619173633","_number":737941,"virtual_id_number":737941,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-25 10:01:24.000000000","_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-25 10:27:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":2,"date":"2020-06-25 10:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"value":1,"date":"2020-06-25 10:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 10:12:23.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-25 10:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7a1df0e05538ad8c5104d0e4395d388282fc866e","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2020-06-25 09:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9fdfc040359a135a4c6b440c217d5d43ca02b18f","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2020-06-25 10:01:24.000000000","message":"Patch Set 1:\n\nI think the heat thing is good now, but I am going to keep a close eye on it. Scaling back up to 20 and monitoring","accounts_in_message":[],"_revision_number":1},{"id":"4fccc7703f89ea865421b9e7b8bfc2a60f0dad5c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 10:11:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9886e88cdcf19538762aa622c6b6ebc24ec64d49","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 10:12:23.000000000","message":"Patch Set 1: Workflow+1\n\nsingle core approving cloud scale-up","accounts_in_message":[],"_revision_number":1},{"id":"a4981dd958884010605e1af89cdc04cce7c59dc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64f7bd2eea8c46858295e0e183491a9c : SUCCESS in 7m 52s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/e4999bfa1eea444db3fe9746c060a1fc : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":1},{"id":"42221b7c1d49d30c858c233537261ea78bacf86b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:13:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75f92b34a4cce9860443c6d9d5c4d8a12ccb6ffe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:27:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e86288859f1f4e5a819b61a5d448c2a4 : SUCCESS in 8m 30s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/2bb11672983949cfadfec959359fe7ff : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"984113ba83a92b52ae5ec4afb5c4ab7d9c98535b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:27:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"042fc62204532905fc4819942b60bff619173633","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:38:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/8dfc2b1b4abb421789c43530b0685b3a : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b90bd45a9684ea6108160dedd1c5d5222d43d17","revisions":{"8b90bd45a9684ea6108160dedd1c5d5222d43d17":{"kind":"REWORK","_number":1,"created":"2020-06-25 09:59:41.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/41/737941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/41/737941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/737941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/737941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/41/737941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/41/737941/1"}}},"commit":{"parents":[{"commit":"bdd09162fdbe30066ad098cd0943d5032ace3ede","subject":"Merge \"grafyaml: drop python2 jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bdd09162fdbe30066ad098cd0943d5032ace3ede"}]}],"author":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2020-06-25 09:59:32.000000000","tz":-240},"committer":{"name":"Donny Davis","email":"donny@fortnebula.com","date":"2020-06-25 09:59:40.000000000","tz":-240},"subject":"Slowly Scale OE back up","message":"Slowly Scale OE back up\n\nChange-Id: Ifec9058bda07c9f50fb3d7178067142828f4ce33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b90bd45a9684ea6108160dedd1c5d5222d43d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8b90bd45a9684ea6108160dedd1c5d5222d43d17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
