)]}'
{"id":"openstack%2Fproject-config~738030","triplet_id":"openstack%2Fproject-config~master~Iaecc9d4fa39b388f9f88729ffdd89bc5e431bc9d","project":"openstack/project-config","branch":"master","topic":"complete-retirement-networking-l2gw","hashtags":[],"change_id":"Iaecc9d4fa39b388f9f88729ffdd89bc5e431bc9d","subject":"Retire networking-l2gw and networking-l2gw-tempest-plugin","status":"MERGED","created":"2020-06-25 13:57:40.000000000","updated":"2020-08-17 13:36:31.000000000","submitted":"2020-08-17 13:36:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738030-1597671391253-c5357c5d","meta_rev_id":"c953527796a1b02a4a7deb471a9c1cdf1e305309","_number":738030,"virtual_id_number":738030,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 13:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-17 06:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-08-17 13:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-08-16 01:03:24.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-08-17 13:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-17 06:50:22.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-17 13:26:07.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-08-17 13:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b743283605681907e66c2ff3fb47a75dba7400f1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 13:57:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"859a27782aad8395925ef512a55fe0d9c0e317ca","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 14:10:38.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"956d26125bf570a53b669313ea54607ebe3f2727","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 14:16:13.000000000","message":"Patch Set 2:\n\nI see that there are recent changes to the repos, could you send an email to the openstack-discuss list to point this out, please?\n\nIf anybody wants to import the repo in another namespace, they can do - let\u0027s tell them and not surprise too much...","accounts_in_message":[],"_revision_number":2},{"id":"8cbfb43876b604c03f3f987b56aa2ae9ced06c46","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 14:19:44.000000000","message":"Patch Set 2: Workflow-1\n\nCheck:\nhttp://codesearch.openstack.org/?q\u003dopenstack%2Fnetworking-l2gw\u0026i\u003dnope\u0026files\u003d\u0026repos\u003d\n\nPlease remove networking-l2gw from all Zuul config files, especially  from networking-midonet, networking-odl, openstack-zuul-jobs, vmware-nsx, requirements, tempest.\nLet\u0027s get those changes merged first to not break these projects.","accounts_in_message":[],"_revision_number":2},{"id":"4e6d9690c4e80089c6c20ecb76fceb4ff0a91f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:28:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6e754b207d0245b8a88b7ac8b538f1c8 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":2},{"id":"481e4c4e71d193bed348030a64dc5fcd180cbf89","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-25 14:42:03.000000000","message":"Patch Set 2:\n\nAnnouncement:\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html","accounts_in_message":[],"_revision_number":2},{"id":"d649d46344f38b3f4c4fd6b9dfcd648d494a7aa2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-25 14:44:04.000000000","message":"Patch Set 2:\n\n\u003e Announcement:\n \u003e \n \u003e http://lists.openstack.org/pipermail/openstack-discuss/2020-June/015647.html\n\nLet\u0027s see if any dependency which can be solved easily or need more time.","accounts_in_message":[],"_revision_number":2},{"id":"bbf30771c58e728d8681dc5d7ed066292fe4abb6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-06-26 15:05:05.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ed6e8816d651b518787fded5c7eb5660602c92d0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 17:25:12.000000000","message":"Patch Set 2: -Workflow\n\nlet\u0027s merge this to proceed on https://review.opendev.org/#/c/745161/\n\nPlan is:\nhttps://etherpad.opendev.org/p/tc-retirement-cleanup L35","accounts_in_message":[],"_revision_number":2},{"id":"5437551c1a8da3560f875be27d5ec561a9266e6a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-14 19:27:45.000000000","message":"Patch Set 2:\n\nare all changes up for this? I\u0027d like to review the complete series first.","accounts_in_message":[],"_revision_number":2},{"id":"bf0e58b6820d5e0f5d18a93bf3e0d797cefe2cc7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-14 23:37:37.000000000","message":"Patch Set 2:\n\n\u003e are all changes up for this? I\u0027d like to review the complete series\n \u003e first.\n\nits here https://review.opendev.org/#/q/topic:complete-retirement-networking-l2gw+(status:open+OR+status:merged)","accounts_in_message":[],"_revision_number":2},{"id":"0ddac6894ca6997764cc4580208b6c85e9590f14","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-15 18:07:55.000000000","message":"Patch Set 2: -Workflow\n\nThe etherpad shows that this is step 1.2, let\u0027s get step 1.1 merged first.\n\nI\u0027ll +2 once this is done.","accounts_in_message":[],"_revision_number":2},{"id":"00138f3c3619415705e453b8e4abdfaf5ceea5e5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-16 01:03:24.000000000","message":"Patch Set 2:\n\n\u003e The etherpad shows that this is step 1.2, let\u0027s get step 1.1 merged\n \u003e first.\n \u003e \n \u003e I\u0027ll +2 once this is done.\n\nI have moved the 1.1 step as 6th, As we are importing the repo first and then retire from openstack, so let\u0027s switch the current usage to new location than removing it now and re-adding it with new location.","accounts_in_message":[],"_revision_number":2},{"id":"0ce554c4eaf47cfc8fd7b8ee65aa2ec68faee512","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-17 06:50:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c7f4c0748d7ebf80ce13a43292a1c308f2696732","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-17 13:26:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"adf43f6d6abd1ad1fd6f5724a6e7588ffb3bfd2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 13:26:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"75bf87022db767dc3a079b0df5c50b8c1e6a0fc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 13:36:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/074de1ea570b4ebca73ba62e9b14beaa : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4997c991a9ff1ba979f24d1af0b92e02ee3ad0e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 13:36:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2ff0b4e120d11bcf81e593198900996ddcb646f","revisions":{"3f5b735cea145a5c3c535fa36db8a61494384547":{"kind":"REWORK","_number":1,"created":"2020-06-25 13:57:40.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/30/738030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/738030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/738030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/738030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/738030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/738030/1"}}},"commit":{"parents":[{"commit":"f937bf0f71c57156ef09d52948312148b6beedd6","subject":"Merge \"Removing missed tripleo-ui references\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f937bf0f71c57156ef09d52948312148b6beedd6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 13:53:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 13:53:03.000000000","tz":-300},"subject":"Retire networking-l2gw and networking-l2gw-tempest-plugin","message":"Retire networking-l2gw and networking-l2gw-tempest-plugin\n\nnetworking-l2gw has been moved from neutron stadium projects\nin ocata and also marked as retired\n- https://review.opendev.org/#/c/392010/2\n\nand networking-l2gw-tempest-plugin was never a official project.\n\nIn disucssion with neutron team on IRC, there is no plan or active\nteam for this projects under neutron stadum so it is good to complete\nthe retirement of those repo.\n\n- http://eavesdrop.openstack.org/irclogs/%23openstack-tc/%23openstack-tc.2020-06-24.log.html#t2020-06-24T15:31:31\n\nThis commit adds the noop job got their gate to start the repo cleanup\n\nChange-Id: Iaecc9d4fa39b388f9f88729ffdd89bc5e431bc9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f5b735cea145a5c3c535fa36db8a61494384547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3f5b735cea145a5c3c535fa36db8a61494384547"}]},"branch":"refs/heads/master"},"a2ff0b4e120d11bcf81e593198900996ddcb646f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-25 14:10:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/30/738030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/30/738030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/738030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/738030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/30/738030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/30/738030/2"}}},"commit":{"parents":[{"commit":"f937bf0f71c57156ef09d52948312148b6beedd6","subject":"Merge \"Removing missed tripleo-ui references\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f937bf0f71c57156ef09d52948312148b6beedd6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 13:53:03.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-06-25 14:10:36.000000000","tz":0},"subject":"Retire networking-l2gw and networking-l2gw-tempest-plugin","message":"Retire networking-l2gw and networking-l2gw-tempest-plugin\n\nnetworking-l2gw has been moved from neutron stadium projects\nin ocata and also marked as retired\n- https://review.opendev.org/#/c/392010/2\n\nand networking-l2gw-tempest-plugin was never a official project.\n\nIn disucssion with neutron team on IRC, there is no plan or active\nteam for this projects under neutron stadum so it is good to complete\nthe retirement of those repo.\n\n- http://eavesdrop.openstack.org/irclogs/%23openstack-tc/%23openstack-tc.2020-06-24.log.html#t2020-06-24T15:31:31\n\nThis commit adds the noop job got their gate to start the repo cleanup\n\nNeeded-By: https://review.opendev.org/#/c/738038/\nNeeded-By: https://review.opendev.org/#/c/738035/\nChange-Id: Iaecc9d4fa39b388f9f88729ffdd89bc5e431bc9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2ff0b4e120d11bcf81e593198900996ddcb646f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a2ff0b4e120d11bcf81e593198900996ddcb646f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
