)]}'
{"id":"openstack%2Fproject-config~738168","triplet_id":"openstack%2Fproject-config~master~I76b5ad7eacd9494cbf22223074796ab4e5599c5c","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I76b5ad7eacd9494cbf22223074796ab4e5599c5c","subject":"Removes py35, tox and cover jobs for proliantutils","status":"MERGED","created":"2020-06-26 09:09:03.000000000","updated":"2020-06-26 14:59:30.000000000","submitted":"2020-06-26 14:59:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738168-1593183570310-995c8465","meta_rev_id":"7d629c6130aeb93eb9ea153d01c5bfd694536bbf","_number":738168,"virtual_id_number":738168,"owner":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 14:59:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2020-06-26 14:30:38.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2020-06-26 14:45:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2020-06-26 14:42:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":1,"date":"2020-06-26 14:45:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 14:30:38.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-06-26 14:42:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-06-26 14:45:33.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-06-26 14:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea7ca5c8e278aa750524e6dabfbd4206d648ad17","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-06-26 09:09:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70408e5d80ec4b0e8df306affe5e140e538082fb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-26 09:13:37.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nlet\u0027s move this into your own repo instead, please.","accounts_in_message":[],"_revision_number":1},{"id":"afa45659b57f5143b90c6681e16f4229354e8a16","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-06-26 09:14:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4d1ada4000e4455bb9829350befa29921636c373","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-26 09:16:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ceecff50c3a786a4d0d700e6fb2e90af1b14990e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-26 09:16:44.000000000","message":"Patch Set 2: Code-Review-1\n\nPlease update commit message for the additional changes","accounts_in_message":[],"_revision_number":2},{"id":"ed41fc692b2fac647be56e9faf5e481ba5099410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 09:29:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eab6d71ddcfa41afa865e31e5260d0b8 : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":2},{"id":"cb234221fb6bdcba1ebe85db7678e93bc9eb4d60","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-06-26 14:22:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"17278c960f4921e5c8c2b064b0519ee202d030fa","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-06-26 14:22:51.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThank you Andreas.","accounts_in_message":[],"_revision_number":2},{"id":"eab9dbd51aa4797aa137812707478281c0cb5fc0","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-06-26 14:30:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2ce368a366a30c4b505ef1b4ee8716bb830f989a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:35:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/83a590db9baa4c7b81c54dd7574b9ad3 : SUCCESS in 7m 11s","accounts_in_message":[],"_revision_number":3},{"id":"302ae4e327307448ff9eaf08379242bc5f6df893","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-26 14:42:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"231aae9e23d86601a270714208d09dd9e7810892","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-26 14:45:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d6752049053de3eae730b51976f6dcc01e6f0d2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:46:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"55ce742ca55347ff75c636983bacf49aaffa04f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:59:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b19e191d84a0443793ac9d6e3432cec6 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":3},{"id":"7d629c6130aeb93eb9ea153d01c5bfd694536bbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:59:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87ba90f67b0e4bbc0345f7b9995d828af350b2ef","revisions":{"fd0626334412bf133a4e31a88280e8d27b91931c":{"kind":"REWORK","_number":1,"created":"2020-06-26 09:09:03.000000000","uploader":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"ref":"refs/changes/68/738168/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/738168/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/738168/1"}}},"commit":{"parents":[{"commit":"d42a4c4dbc1b9cd96b16835245f5ec020b18d9a2","subject":"Merge \"Retire networking-onos, openstack-ux, solum-infra-guest-agent: Step 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d42a4c4dbc1b9cd96b16835245f5ec020b18d9a2"}]}],"author":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2020-06-26 10:56:54.000000000","tz":-240},"committer":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2020-06-26 10:58:00.000000000","tz":-240},"subject":"Removes py35 and py27 jobs for proliantutils","message":"Removes py35 and py27 jobs for proliantutils\n\nChange-Id: I76b5ad7eacd9494cbf22223074796ab4e5599c5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd0626334412bf133a4e31a88280e8d27b91931c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd0626334412bf133a4e31a88280e8d27b91931c"}]},"branch":"refs/heads/master"},"18d86d899b7dbe02b79e9ce00871b53c6e0593b7":{"kind":"REWORK","_number":2,"created":"2020-06-26 09:14:51.000000000","uploader":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"ref":"refs/changes/68/738168/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/738168/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/738168/2"}}},"commit":{"parents":[{"commit":"d42a4c4dbc1b9cd96b16835245f5ec020b18d9a2","subject":"Merge \"Retire networking-onos, openstack-ux, solum-infra-guest-agent: Step 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d42a4c4dbc1b9cd96b16835245f5ec020b18d9a2"}]}],"author":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2020-06-26 10:56:54.000000000","tz":-240},"committer":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2020-06-26 11:03:32.000000000","tz":-240},"subject":"Removes py35 and py27 jobs for proliantutils","message":"Removes py35 and py27 jobs for proliantutils\n\nChange-Id: I76b5ad7eacd9494cbf22223074796ab4e5599c5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18d86d899b7dbe02b79e9ce00871b53c6e0593b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/18d86d899b7dbe02b79e9ce00871b53c6e0593b7"}]},"branch":"refs/heads/master"},"87ba90f67b0e4bbc0345f7b9995d828af350b2ef":{"kind":"REWORK","_number":3,"created":"2020-06-26 14:22:41.000000000","uploader":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"ref":"refs/changes/68/738168/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/68/738168/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/68/738168/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/68/738168/3"}}},"commit":{"parents":[{"commit":"d42a4c4dbc1b9cd96b16835245f5ec020b18d9a2","subject":"Merge \"Retire networking-onos, openstack-ux, solum-infra-guest-agent: Step 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d42a4c4dbc1b9cd96b16835245f5ec020b18d9a2"}]}],"author":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2020-06-26 10:56:54.000000000","tz":-240},"committer":{"name":"Shivanand Tendulker","email":"stendulker@gmail.com","date":"2020-06-26 16:10:17.000000000","tz":-240},"subject":"Removes py35, tox and cover jobs for proliantutils","message":"Removes py35, tox and cover jobs for proliantutils\n\nChange-Id: I76b5ad7eacd9494cbf22223074796ab4e5599c5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87ba90f67b0e4bbc0345f7b9995d828af350b2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/87ba90f67b0e4bbc0345f7b9995d828af350b2ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
