)]}'
{"id":"openstack%2Fproject-config~738378","triplet_id":"openstack%2Fproject-config~master~I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe","project":"openstack/project-config","branch":"master","topic":"tobiko-devstack","hashtags":[],"change_id":"I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe","subject":"Create a new repository for Tobiko DevStack plugin","status":"MERGED","created":"2020-06-29 06:13:18.000000000","updated":"2020-07-02 10:22:35.000000000","submitted":"2020-07-02 05:52:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738378-1593669146408-f3086ce0","meta_rev_id":"5cbb8c35cf09b83fccd4c08a43bd88db1e027a08","_number":738378,"virtual_id_number":738378,"owner":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-02 05:52:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":2,"date":"2020-07-01 05:48:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-07-01 09:52:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-02 05:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-02 05:43:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-29 12:31:01.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-01 09:52:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-02 05:43:59.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-07-02 05:52:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab3d2bc2f60d5f2dbee40a7ec377aae78d594e5f","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 06:13:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bc266e87d1e809dd18fa44d45e8f5019e0e1ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 06:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/887d63ebbe8c41d0aa91d5024c781cfc : SUCCESS in 8m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d4a047e223514dcf8de7d74823abea66 : SUCCESS in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9c441f57dc6ef53cf851f0318d321e4a211b3e2c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-29 06:56:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nDo you want to import initial content with history?","accounts_in_message":[],"_revision_number":1},{"id":"cb92f39e5fbdf520279b3408e0869535acb05548","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 07:36:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b362c4ccae1fb7cb0727844c83c8678a560ea547","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 07:36:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c883061659ebb9ae7e897ffda5de74e2f6f79bdf","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 07:36:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75bd135a1f25d3ea966356f0ad2b5ad93ba3f95d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-29 07:39:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8857054f5aebc092c930689c44e00e292869c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 07:49:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8693532b540b4cfb848cefdbf3aa8614 : SUCCESS in 8m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/31f42a454102494db10be400911f5b07 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":3},{"id":"64cbbb10bef51a86f96f36313bebbfa6db9457f3","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 08:02:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"275303e244ba0b918b099065130c012d49e17ceb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:13:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a2dc82bbeef490d94f2e0be3c712064 : SUCCESS in 9m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/07fae21a097a4f7db58952078ee349f0 : FAILURE in 5m 15s","accounts_in_message":[],"_revision_number":4},{"id":"21f1755577b4364aa21884598bbc8c82fec4ca84","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 08:27:47.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"fc3d21aa7b30cd6a5858dc8df186a55a08352e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 08:36:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/96bf778424204564b2358205033fd7ff : SUCCESS in 7m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8661f2a648024333bce3ca5742180a03 : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":5},{"id":"d0b60416d24856df0340bdab77d6d201065dc3bc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-29 08:39:19.000000000","message":"Patch Set 5:\n\nLet me ask again:\n\n \u003e Do you want to import initial content with history?\n\nNote that you will start with an empty repository. Your very first\npatch needs to setup the repository and add all infrastructure so that\nthe jobs pass (this can be done with a cookiecutter project).\n\nAlso, you can import an existing repository including it\u0027s history if\nyou have one. Once the repository is created, importing is not\npossible anymore, changes will have to go through the normal review\nprocess.\n\nIf you have a lot of history to import, please use the upstream field\ninstead of creating a repository and then pushing the patches one at a\ntime. Pushing a large number of related patches all at one time causes\nthe CI infrastructure to slow down, which impacts work on all of the\nother projects using it. After repo creation the infra team will not\npush any content to it; this is your only chance to import directly.\n\nPlease confirm that starting with an empty repo is fine.","accounts_in_message":[],"_revision_number":5},{"id":"e882da9ced073b3dbb814c97fa078867e586bb11","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-06-29 12:14:07.000000000","message":"Patch Set 5:\n\nHello Andreas,\n\nsorry I forgot to mention. For me it is fine starting a repository from scratch. I don\u0027t have any history to keep.\n\n| Your very first patch needs to setup the repository and add all \n| infrastructure so that the jobs pass (this can be done with a\n| cookiecutter project).\n\nI am interested in knowing more regarding these first steps.\n\n| Please confirm that starting with an empty repo is fine.\n\nI confirm you it is fine. I will submit some incremental patches to it copying code (not much) from another repo, without caring about its history.\n\nThank you","accounts_in_message":[],"_revision_number":5},{"id":"410b3b8d11b450e5c4b8f12f615bd9c65d908a39","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-06-29 12:31:01.000000000","message":"Patch Set 5: Code-Review+2\n\n\u003e Hello Andreas,\n \u003e \n \u003e sorry I forgot to mention. For me it is fine starting a repository\n \u003e from scratch. I don\u0027t have any history to keep.\n\n\ngood, then this is fine as is.\n\n \u003e | Your very first patch needs to setup the repository and add all\n \u003e | infrastructure so that the jobs pass (this can be done with a\n \u003e | cookiecutter project).\n \u003e \n \u003e I am interested in knowing more regarding these first steps.\n\nJust add a .zuul.yaml file with either noop-jobs (for both check and gate) or add real jobs, happy to review a change if you run into problems.","accounts_in_message":[],"_revision_number":5},{"id":"33fb2055b02f1f38f9c54e2d4d4b29959d560aff","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-01 05:48:09.000000000","message":"Patch Set 5:\n\nThank you very much Andreas!","accounts_in_message":[],"_revision_number":5},{"id":"a966b243810d08b66b871d6a91e10c2dbcb1fc29","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-01 05:48:19.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3b8a3899dd663c0c6ee3c2f950466b277f235c64","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-01 05:50:55.000000000","message":"Patch Set 6:\n\nFederico, why did you rebase? That is not needed at all - gerrit automatically rebases before test/merge.","accounts_in_message":[],"_revision_number":6},{"id":"6046a37b80c3eec8fab92d08549b7094a2c234eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 05:57:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/189becb416914e9db723fd03b9fa212c : SUCCESS in 8m 24s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8917264f26504bdf8d472ae41ae92a06 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":6},{"id":"e1682fbf59f8462f9a4104768b722af78209e180","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-01 09:52:14.000000000","message":"Patch Set 6: Code-Review+1\n\nHi Federico,\n\nNice to see tobiko here.\nDo not hesitate to ping me if you need on the releases and associated processes.\n\nCheers","accounts_in_message":[],"_revision_number":6},{"id":"88ce725f1298559e1514066f0d9300ab25328e39","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-02 05:43:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"dcb1e9cc38e3b0fa4b4663cf06a3b38769a17e78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:44:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"188521b1bc1bb0fe14c50d38521bc6ebad136925","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:52:26.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cb04c10a1a8b4d4ba25e8d6893838a05 : SUCCESS in 7m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6bb51a98c924480191bea03d85872600 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":6},{"id":"a7e0ea625e9098550405daa67f58f26737b52b36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 05:52:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"81acdb4b8ce485b5c5fab23a44ce18a5a2927745","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 09:33:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/2406546b516a4ba3bf79de0c4ee2447e : SUCCESS in 3m 18s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/870bb3a8e3a54bdbbd97d0060316b1ec : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":6},{"id":"0be520d08e920f7c8017b5aeaa5e17f1d3f8eb73","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:22:35.000000000","message":"Created a revert of this change as I77b91dd03a28f8960d7294c1ef086f4b46b02ccd","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c7ca7592a98b9b9e229f3e73521f5b12a4945a2e","revisions":{"de3848f66a94704d51149b9349d144cb3e7b1824":{"kind":"REWORK","_number":1,"created":"2020-06-29 06:13:18.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/78/738378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/738378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/738378/1"}}},"commit":{"parents":[{"commit":"ec84007c08b22f5496966f8caa40fb809b24d520","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec84007c08b22f5496966f8caa40fb809b24d520"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:05:02.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:13:15.000000000","tz":120},"subject":"Create a new repository for Tobiko DevStack plugin","message":"Create a new repository for Tobiko DevStack plugin\n\nTobiko project repo (x/tobiko) grown fast since past few years\nand from since a simple Python framework adquired additional\ncomponents that could be considered separate sub projects.\n\nOne of them is its DevStack plugin. As Tobiko can today connect\nto an DevStack deployed cloud without requiring any configuration\nthis plugin has become optional and there this change serves\nto the need of migrating it to its own location.\n\nThis new repo should also serve as location for DevStack related\nZuul jobs that are intended for running Tobiko test cases\nagainst a DevStack provided cloud on OpenDev CI.\n\nIt this way we would also remove Zuul\u0027s dependency of Tobiko\nmain project jobs from DevStack project jobs helping making\na more granular an rational use of CI resources at the time\nof submitting new changes to Tobiko main repo itself by triggering\nbuild only when changing the framework or test cases, while\nignorin the rest of related dayly changes.\n\nChange-Id: I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de3848f66a94704d51149b9349d144cb3e7b1824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/de3848f66a94704d51149b9349d144cb3e7b1824"}]},"branch":"refs/heads/master"},"3e20f0d6d84aa4a2299541238a123335aeda0e1e":{"kind":"REWORK","_number":2,"created":"2020-06-29 07:36:01.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/78/738378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/738378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/738378/2"}}},"commit":{"parents":[{"commit":"ec84007c08b22f5496966f8caa40fb809b24d520","subject":"Merge \"Normalize projects.yaml\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec84007c08b22f5496966f8caa40fb809b24d520"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:05:02.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 07:35:56.000000000","tz":0},"subject":"Create a new repository for Tobiko DevStack plugin","message":"Create a new repository for Tobiko DevStack plugin\n\nTobiko project repo (x/tobiko) grown fast since past few years\nand from since a simple Python framework adquired additional\ncomponents that could be considered separate sub projects.\n\nOne of them is its DevStack plugin. As Tobiko can today connect\nto an DevStack deployed cloud without requiring any configuration\nthis plugin has become optional and there this change serves\nto the need of migrating it to its own location.\n\nThis new repo should also serve as location for DevStack related\nZuul jobs that are intended for running Tobiko test cases\nagainst a DevStack provided cloud on OpenDev CI.\n\nIt this way we would also remove Zuul\u0027s dependency of Tobiko\nmain project jobs from DevStack project jobs helping making\na more granular an rational use of CI resources at the time\nof submitting new changes to Tobiko main repo itself by triggering\nbuild only when changing the framework or test cases, while\nignorin the rest of related dayly changes.\n\nChange-Id: I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e20f0d6d84aa4a2299541238a123335aeda0e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3e20f0d6d84aa4a2299541238a123335aeda0e1e"}]},"branch":"refs/heads/master"},"d9db0df288afeb7a863cdaa5133520f069358864":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-29 07:36:07.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/78/738378/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/738378/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/738378/3"}}},"commit":{"parents":[{"commit":"d98d0996bb60ffac845d85108e6b6ec60fff2825","subject":"Merge \"Finish retirement of openstack-ux,solum-infra-guestagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d98d0996bb60ffac845d85108e6b6ec60fff2825"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:05:02.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 07:36:07.000000000","tz":0},"subject":"Create a new repository for Tobiko DevStack plugin","message":"Create a new repository for Tobiko DevStack plugin\n\nTobiko project repo (x/tobiko) grown fast since past few years\nand from since a simple Python framework adquired additional\ncomponents that could be considered separate sub projects.\n\nOne of them is its DevStack plugin. As Tobiko can today connect\nto an DevStack deployed cloud without requiring any configuration\nthis plugin has become optional and there this change serves\nto the need of migrating it to its own location.\n\nThis new repo should also serve as location for DevStack related\nZuul jobs that are intended for running Tobiko test cases\nagainst a DevStack provided cloud on OpenDev CI.\n\nIt this way we would also remove Zuul\u0027s dependency of Tobiko\nmain project jobs from DevStack project jobs helping making\na more granular an rational use of CI resources at the time\nof submitting new changes to Tobiko main repo itself by triggering\nbuild only when changing the framework or test cases, while\nignorin the rest of related dayly changes.\n\nChange-Id: I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9db0df288afeb7a863cdaa5133520f069358864"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9db0df288afeb7a863cdaa5133520f069358864"}]},"branch":"refs/heads/master"},"148caabc26dbfdc5dbecdd68d1e6a41f58e3cf12":{"kind":"REWORK","_number":4,"created":"2020-06-29 08:02:12.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/78/738378/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/738378/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/738378/4"}}},"commit":{"parents":[{"commit":"d98d0996bb60ffac845d85108e6b6ec60fff2825","subject":"Merge \"Finish retirement of openstack-ux,solum-infra-guestagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d98d0996bb60ffac845d85108e6b6ec60fff2825"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:05:02.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 08:01:56.000000000","tz":120},"subject":"Create a new repository for Tobiko DevStack plugin","message":"Create a new repository for Tobiko DevStack plugin\n\nTobiko project repo (x/tobiko) grown fast since past few years\nand from since a simple Python framework adquired additional\ncomponents that could be considered separate sub projects.\n\nOne of them is its DevStack plugin. As Tobiko can today connect\nto an DevStack deployed cloud without requiring any configuration\nthis plugin has become optional and there this change serves\nto the need of migrating it to its own location.\n\nThis new repo should also serve as location for DevStack related\nZuul jobs that are intended for running Tobiko test cases\nagainst a DevStack provided cloud on OpenDev CI.\n\nIt this way we would also remove Zuul\u0027s dependency of Tobiko\nmain project jobs from DevStack project jobs helping making\na more granular an rational use of CI resources at the time\nof submitting new changes to Tobiko main repo itself by triggering\nbuild only when changing the framework or test cases, while\nignorin the rest of related dayly changes.\n\nChange-Id: I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/148caabc26dbfdc5dbecdd68d1e6a41f58e3cf12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/148caabc26dbfdc5dbecdd68d1e6a41f58e3cf12"}]},"branch":"refs/heads/master"},"2658cb1c51cd9b2ab3c462d6393d3ab2e06fc15f":{"kind":"REWORK","_number":5,"created":"2020-06-29 08:27:47.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/78/738378/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/738378/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/738378/5"}}},"commit":{"parents":[{"commit":"d98d0996bb60ffac845d85108e6b6ec60fff2825","subject":"Merge \"Finish retirement of openstack-ux,solum-infra-guestagent\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d98d0996bb60ffac845d85108e6b6ec60fff2825"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:05:02.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 08:27:41.000000000","tz":0},"subject":"Create a new repository for Tobiko DevStack plugin","message":"Create a new repository for Tobiko DevStack plugin\n\nTobiko project repo (x/tobiko) grown fast since past few years\nand from since a simple Python framework adquired additional\ncomponents that could be considered separate sub projects.\n\nOne of them is its DevStack plugin. As Tobiko can today connect\nto an DevStack deployed cloud without requiring any configuration\nthis plugin has become optional and there this change serves\nto the need of migrating it to its own location.\n\nThis new repo should also serve as location for DevStack related\nZuul jobs that are intended for running Tobiko test cases\nagainst a DevStack provided cloud on OpenDev CI.\n\nIt this way we would also remove Zuul\u0027s dependency of Tobiko\nmain project jobs from DevStack project jobs helping making\na more granular an rational use of CI resources at the time\nof submitting new changes to Tobiko main repo itself by triggering\nbuild only when changing the framework or test cases, while\nignorin the rest of related dayly changes.\n\nChange-Id: I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2658cb1c51cd9b2ab3c462d6393d3ab2e06fc15f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2658cb1c51cd9b2ab3c462d6393d3ab2e06fc15f"}]},"branch":"refs/heads/master"},"c7ca7592a98b9b9e229f3e73521f5b12a4945a2e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-07-01 05:48:19.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/78/738378/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/738378/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/738378/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/738378/6"}}},"commit":{"parents":[{"commit":"ac49fc6967e4087ef361222e15ce80f04f5416f6","subject":"Merge \"Finish retirement of networking-onos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac49fc6967e4087ef361222e15ce80f04f5416f6"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-06-29 06:05:02.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-01 05:48:19.000000000","tz":0},"subject":"Create a new repository for Tobiko DevStack plugin","message":"Create a new repository for Tobiko DevStack plugin\n\nTobiko project repo (x/tobiko) grown fast since past few years\nand from since a simple Python framework adquired additional\ncomponents that could be considered separate sub projects.\n\nOne of them is its DevStack plugin. As Tobiko can today connect\nto an DevStack deployed cloud without requiring any configuration\nthis plugin has become optional and there this change serves\nto the need of migrating it to its own location.\n\nThis new repo should also serve as location for DevStack related\nZuul jobs that are intended for running Tobiko test cases\nagainst a DevStack provided cloud on OpenDev CI.\n\nIt this way we would also remove Zuul\u0027s dependency of Tobiko\nmain project jobs from DevStack project jobs helping making\na more granular an rational use of CI resources at the time\nof submitting new changes to Tobiko main repo itself by triggering\nbuild only when changing the framework or test cases, while\nignorin the rest of related dayly changes.\n\nChange-Id: I8ec33cefd5a83fe111d4d9b58c4088ef4b990dbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7ca7592a98b9b9e229f3e73521f5b12a4945a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7ca7592a98b9b9e229f3e73521f5b12a4945a2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
