)]}'
{"id":"openstack%2Fproject-config~738979","triplet_id":"openstack%2Fproject-config~master~Ic8fb712534dc064190bd372a47f6217079cf0554","project":"openstack/project-config","branch":"master","topic":"project-rename","hashtags":[],"change_id":"Ic8fb712534dc064190bd372a47f6217079cf0554","subject":"Fix x/devstack-plugin-tobiko name","status":"MERGED","created":"2020-07-02 08:03:54.000000000","updated":"2020-07-24 15:45:36.000000000","submitted":"2020-07-24 15:37:57.000000000","submitter":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"738979-1595605077447-5004bf0a","meta_rev_id":"a26408bda3287912e399b610576d6e180710bc08","_number":738979,"virtual_id_number":738979,"owner":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":2,"date":"2020-07-24 15:37:52.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-04 09:07:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":2,"date":"2020-07-23 19:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-07-23 19:49:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":1,"date":"2020-07-24 15:37:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"date":"2020-07-23 19:49:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-04 09:07:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-23 19:49:53.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-24 15:37:52.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"f4e62cd5bbae3dff6b4fb8bfcd9f977725006e4b","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 08:03:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44f1cff0c0b89a1bb64b2691ce4fbf5cef53c125","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-02 08:12:53.000000000","message":"Patch Set 1: Workflow-1\n\nArgh, this needs a project rename - we cannot easily change it this way ;(\n\nSee https://docs.opendev.org/opendev/infra-manual/latest/creators.html#project-renames and follow the process there.","accounts_in_message":[],"_revision_number":1},{"id":"7de67eaf1f7ce4a0eb1d2b13fef7ba62e3e3721e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 08:16:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6f5ca92b93c3415592604e7f990300b3 : SUCCESS in 9m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/389bf603e2db4a8bbabce641f02ef6bb : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"0f3bfc7797fa43cffb2351723975c86bf686b0b8","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 08:44:00.000000000","message":"Patch Set 1:\n\nPlease note I never pushed changes to the old repo name. It should be still empty I guess.","accounts_in_message":[],"_revision_number":1},{"id":"6e8cb560657fb7bfe4c73878e16bb23ec448795a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-02 08:53:55.000000000","message":"Patch Set 1:\n\nWe cannot easily rename ;( Please use the repo.","accounts_in_message":[],"_revision_number":1},{"id":"5c0a512b00a6d2f536c675fde9797a0f4ad7edac","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:16:43.000000000","message":"Patch Set 1:\n\nCan we please remove existing repo then re-add the new one? The repo is empty so it should be not an issue doing it.","accounts_in_message":[],"_revision_number":1},{"id":"f1c98e48e708824e290fe8f30743bb15e8219b39","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:19:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d927fe6af9caf8fc1fd1e3fafbfb4b05ba194809","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-02 10:19:56.000000000","message":"Patch Set 1:\n\nPlease ask on #opendev channel once the US based admins are awake.\n\nI cannot help further.","accounts_in_message":[],"_revision_number":1},{"id":"08e0e2002294e8dcc58ca9a161cb7fb64cae96fe","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:22:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"97f3afd6d0650ed7b45bf7457201ae07baef8f99","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:23:41.000000000","message":"Topic changed from tobiko to devstack-plugin-tobiko","accounts_in_message":[],"_revision_number":3},{"id":"1cca5696bb98582059a9d0ebf93460c936405297","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-02 10:30:18.000000000","message":"Patch Set 3: Code-Review-2\n\nThis does not work with infrastructure. Please discuss on #opendev as mentioned earlier.","accounts_in_message":[],"_revision_number":3},{"id":"22a5f15c9bc00721fb9945b284bd9e34ff7ecc7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 10:37:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/16a4813989b0453fb57398df74fe895c : SUCCESS in 11m 20s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7c94e665ed7b4c3199d47eaee55814a2 : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":3},{"id":"f40d4d8c901054701d9a01b2b5abf3f2cc933b9e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-02 15:09:10.000000000","message":"Patch Set 3:\n\nFederico, please use the created repo and follow the rename process.","accounts_in_message":[],"_revision_number":3},{"id":"9a682ead6f678a5becfd717598b906c5bb55e98d","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 15:24:47.000000000","message":"Patch Set 3:\n\nBecause the actual content of the repo is still irrelevant (no commits done yet), the rename operation is not required. We could instead create a new one (with the right name) and delete the old one. These two operations can be performed in any order.","accounts_in_message":[],"_revision_number":3},{"id":"9c85c3468f2dd330c6a429f7f9853988badeb004","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 15:25:21.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"1dfe02e0055b1794088d93ee1a9aceef66e7d15b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 15:43:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6964997d5b604f05aaa50ab50843fd54 : SUCCESS in 10m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8228629ce97e49469dd3ee84680d8080 : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":4},{"id":"b86d9c03907ae7150ebd00d55467aaf2efa51b4d","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-03 13:29:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"933bca4b66197941be329dcd186148cbd2e85769","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-03 14:19:02.000000000","message":"Patch Set 4:\n\nFederico, we do need a change for the rename, see my first link... please send one","accounts_in_message":[],"_revision_number":4},{"id":"910fcbdac587b51cb8d10d0b595e08f783d3f098","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-04 08:58:17.000000000","message":"Restored\n\nWe need this change for the rename","accounts_in_message":[],"_revision_number":4},{"id":"a19f3f8aad569b77f70d823fd7763b8d2293046f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-04 09:00:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"98595fc4e6a318457c8667fcb06cf2cd8a59d1b6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-04 09:00:20.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"8ad1be87cccc1621e95ac00b3b0e37890f7a421d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-04 09:02:46.000000000","message":"Patch Set 5:\n\nAdded to our tracking list: https://wiki.openstack.org/wiki/Meetings/InfraTeamMeeting#Upcoming_Project_Renames","accounts_in_message":[],"_revision_number":5},{"id":"6fc06765ffff7aab92d484f23a68a5e2ef07963b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-04 09:02:54.000000000","message":"Topic changed from devstack-plugin-tobiko-patch1 to project-rename","accounts_in_message":[],"_revision_number":5},{"id":"cf9f2e70a5f06bc284daef5e4ce0d805432f95a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 09:07:37.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1bd324c82a1e4f30bdcc9a9712ba0c07 : SUCCESS in 7m 06s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0aa579ae212e45088a0aa24801455586 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":5},{"id":"eb652db5840ac06a85c44ebcf7692b3aff4ae171","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-04 09:09:02.000000000","message":"Patch Set 5: Workflow-1\n\nwaiting for rename","accounts_in_message":[],"_revision_number":5},{"id":"97863248b1088183c1977f3c8e974d30661c3c1f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-23 19:44:32.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"43b0ac6b7b5e336c1fceb55d2a49b5774034d130","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-23 19:49:17.000000000","message":"Patch Set 5: Code-Review+2 -Workflow\n\nPlanned rename window is July 24th 15:00 UTC","accounts_in_message":[],"_revision_number":5},{"id":"442742803fb05d214638e90d4ebd42ce822794d4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-23 19:49:53.000000000","message":"Patch Set 5: Workflow-1\n\nleaving WIP until rename window; infra-root, please remove when ready","accounts_in_message":[],"_revision_number":5},{"id":"83f100ea21b71930a15aee1d5289aa5e47ee4af8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-24 15:37:42.000000000","message":"Removed Workflow-1 by Andreas Jaeger \u003cjaegerandi@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":5},{"id":"0f203dde46921771bcbacaea1cb7bcd8fab0db3a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-24 15:37:52.000000000","message":"Patch Set 5: Verified+2 Workflow+1\n\nForce merging project rename.","accounts_in_message":[],"_revision_number":5},{"id":"eb9e8a7683f3378524ee348823ac23026fb9dd8a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-24 15:37:57.000000000","message":"Change has been successfully merged by Clark Boylan","accounts_in_message":[],"_revision_number":5},{"id":"a26408bda3287912e399b610576d6e180710bc08","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 15:45:36.000000000","message":"Patch Set 5:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/2a507649eb81402192c7b7c72d30b944 : FAILURE in 1m 47s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bd92ad1b8f7979122a278bf9bec00bd68f2b4afd","revisions":{"7d206d0c0e8440286fddef114d26db57d08b09f2":{"kind":"REWORK","_number":1,"created":"2020-07-02 08:03:54.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/79/738979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/738979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/738979/1"}}},"commit":{"parents":[{"commit":"c1a377997dce9c009201091062e1401f285baca6","subject":"Merge \"Create a new project for ansible-tripleo-ipa-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a377997dce9c009201091062e1401f285baca6"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 08:03:39.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 08:03:39.000000000","tz":120},"subject":"Fix x/devstack-plugin-tobiko name","message":"Fix x/devstack-plugin-tobiko name\n\nChange-Id: Ic8fb712534dc064190bd372a47f6217079cf0554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d206d0c0e8440286fddef114d26db57d08b09f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d206d0c0e8440286fddef114d26db57d08b09f2"}]},"branch":"refs/heads/master"},"3d02156e6f4d5da86ae69c8d2984a79689a0ce9c":{"kind":"REWORK","_number":2,"created":"2020-07-02 10:19:49.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/79/738979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/738979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/738979/2"}}},"commit":{"parents":[{"commit":"c1a377997dce9c009201091062e1401f285baca6","subject":"Merge \"Create a new project for ansible-tripleo-ipa-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a377997dce9c009201091062e1401f285baca6"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 08:03:39.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:19:46.000000000","tz":0},"subject":"Fix x/devstack-plugin-tobiko name","message":"Fix x/devstack-plugin-tobiko name\n\nChange-Id: Ic8fb712534dc064190bd372a47f6217079cf0554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d02156e6f4d5da86ae69c8d2984a79689a0ce9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d02156e6f4d5da86ae69c8d2984a79689a0ce9c"}]},"branch":"refs/heads/master"},"782c22f60d0665af1b17a928d93aaf5e329d9b92":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-02 10:22:00.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/79/738979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/738979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/738979/3"}}},"commit":{"parents":[{"commit":"c1a377997dce9c009201091062e1401f285baca6","subject":"Merge \"Create a new project for ansible-tripleo-ipa-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a377997dce9c009201091062e1401f285baca6"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 08:03:39.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:21:48.000000000","tz":0},"subject":"Re-create x/devstack-plugin-tobiko project","message":"Re-create x/devstack-plugin-tobiko project\n\nThe project has been previously created with the wrong name [1]\n\n[1] https://review.opendev.org/738378/\n\nChange-Id: Ic8fb712534dc064190bd372a47f6217079cf0554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/782c22f60d0665af1b17a928d93aaf5e329d9b92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/782c22f60d0665af1b17a928d93aaf5e329d9b92"}]},"branch":"refs/heads/master"},"c7f44d947e61e16f852a01a98d55898ce12853a0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-07-02 15:25:21.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/79/738979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/738979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/738979/4"}}},"commit":{"parents":[{"commit":"b1fae6833bc2093c491b794915e98291e2e26fcb","subject":"Mirror arista and trilio charms to GitHub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1fae6833bc2093c491b794915e98291e2e26fcb"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 08:03:39.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 15:25:21.000000000","tz":0},"subject":"Re-create x/devstack-plugin-tobiko project","message":"Re-create x/devstack-plugin-tobiko project\n\nThe project has been previously created with the wrong name [1]\n\n[1] https://review.opendev.org/738378/\n\nChange-Id: Ic8fb712534dc064190bd372a47f6217079cf0554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7f44d947e61e16f852a01a98d55898ce12853a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7f44d947e61e16f852a01a98d55898ce12853a0"}]},"branch":"refs/heads/master"},"bd92ad1b8f7979122a278bf9bec00bd68f2b4afd":{"kind":"REWORK","_number":5,"created":"2020-07-04 09:00:07.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/79/738979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/738979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/738979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/738979/5"}}},"commit":{"parents":[{"commit":"c1a377997dce9c009201091062e1401f285baca6","subject":"Merge \"Create a new project for ansible-tripleo-ipa-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a377997dce9c009201091062e1401f285baca6"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 08:03:39.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-07-04 08:59:35.000000000","tz":120},"subject":"Fix x/devstack-plugin-tobiko name","message":"Fix x/devstack-plugin-tobiko name\n\nRename repo topiko -\u003e tobiko.\n\nChange-Id: Ic8fb712534dc064190bd372a47f6217079cf0554\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd92ad1b8f7979122a278bf9bec00bd68f2b4afd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd92ad1b8f7979122a278bf9bec00bd68f2b4afd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
