)]}'
{"id":"openstack%2Fproject-config~739006","triplet_id":"openstack%2Fproject-config~master~I77b91dd03a28f8960d7294c1ef086f4b46b02ccd","project":"openstack/project-config","branch":"master","topic":"devstack-plugin-tobiko","hashtags":[],"change_id":"I77b91dd03a28f8960d7294c1ef086f4b46b02ccd","subject":"Revert \"Create a new repository for Tobiko DevStack plugin\"","status":"ABANDONED","created":"2020-07-02 10:22:35.000000000","updated":"2020-07-03 13:30:10.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a2a88bf948d9e9eb0406c85719cd662937022492","_number":739006,"virtual_id_number":739006,"owner":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-02 15:46:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":-2,"date":"2020-07-02 15:25:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-02 10:22:35.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-07-02 10:22:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-02 10:29:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-07-02 15:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45cceaaec218999eac4e1e9670a3faf0793ee412","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:22:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d70b08cbee53c22a61bef680517dc4d2d601596","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:22:47.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3010ffd7a6126fd71d04821dd825759a7b681656","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 10:23:25.000000000","message":"Topic changed from tobiko-devstack to devstack-plugin-tobiko","accounts_in_message":[],"_revision_number":2},{"id":"7140628f87e3442ade706af17455e2cd1c595d39","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-07-02 10:29:47.000000000","message":"Patch Set 2: Code-Review-2\n\nFederico, a revert does not work with infrastructure.\n\nPlease stop it and discuss first in IRC channel #opendev later today.","accounts_in_message":[],"_revision_number":2},{"id":"40d86e1deff751aee4505b0d4fa78d4b2deeae91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 10:37:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4fb8e4ac6e854990b6a6c8908f249e9d : SUCCESS in 8m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3a286d3a403f44a884a2ec7976d11380 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":2},{"id":"ebfd359405775ef1fc02ba4d13dceafc3f9593a5","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 11:11:41.000000000","message":"Patch Set 2:\n\nThank you Andreas","accounts_in_message":[],"_revision_number":2},{"id":"b79bb14a36f8c9178c82233f8be3326f628b6ba7","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 15:25:03.000000000","message":"Patch Set 2:\n\nBecause the actual content of the repo is still irrelevant (no commits done yet), the rename operation is not required. We could instead create a new one (with the right name) and delete the old one. These two operations can be performed in any order.","accounts_in_message":[],"_revision_number":2},{"id":"aa71e43296362c5c21bf7a2dff649f22881e7bbc","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-02 15:25:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1198edb96f5ac0e4bdd8e67e30a3e7f47f36f5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 15:46:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/98bc80e5805748d696c6f5f40b7f17a0 : SUCCESS in 10m 08s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a8504454f3ce4b608d507b0c604552dd : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a2a88bf948d9e9eb0406c85719cd662937022492","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2020-07-03 13:30:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"361e40a8a5301627821108add101f69ff4564e93","revisions":{"7b94f498252625bab1ef263bfae4e8290026936e":{"kind":"REWORK","_number":1,"created":"2020-07-02 10:22:35.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/06/739006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/739006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/739006/1"}}},"commit":{"parents":[{"commit":"c7ca7592a98b9b9e229f3e73521f5b12a4945a2e","subject":"Create a new repository for Tobiko DevStack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c7ca7592a98b9b9e229f3e73521f5b12a4945a2e"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:22:35.000000000","tz":0},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:22:35.000000000","tz":0},"subject":"Revert \"Create a new repository for Tobiko DevStack plugin\"","message":"Revert \"Create a new repository for Tobiko DevStack plugin\"\n\nThis reverts commit c7ca7592a98b9b9e229f3e73521f5b12a4945a2e.\n\nThe project has been created with a mispelled name.\n\nChange-Id: I77b91dd03a28f8960d7294c1ef086f4b46b02ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b94f498252625bab1ef263bfae4e8290026936e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7b94f498252625bab1ef263bfae4e8290026936e"}]},"branch":"refs/heads/master"},"ae19583afb5d3f6a189ab91b724897825e11ee4f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-02 10:22:47.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/06/739006/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/739006/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/739006/2"}}},"commit":{"parents":[{"commit":"c1a377997dce9c009201091062e1401f285baca6","subject":"Merge \"Create a new project for ansible-tripleo-ipa-server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1a377997dce9c009201091062e1401f285baca6"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:22:35.000000000","tz":0},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:22:47.000000000","tz":0},"subject":"Revert \"Create a new repository for Tobiko DevStack plugin\"","message":"Revert \"Create a new repository for Tobiko DevStack plugin\"\n\nThis reverts commit c7ca7592a98b9b9e229f3e73521f5b12a4945a2e.\n\nThe project has been created with a mispelled name.\n\nChange-Id: I77b91dd03a28f8960d7294c1ef086f4b46b02ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae19583afb5d3f6a189ab91b724897825e11ee4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ae19583afb5d3f6a189ab91b724897825e11ee4f"}]},"branch":"refs/heads/master"},"361e40a8a5301627821108add101f69ff4564e93":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-02 15:25:14.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/06/739006/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/06/739006/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/06/739006/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/06/739006/3"}}},"commit":{"parents":[{"commit":"b1fae6833bc2093c491b794915e98291e2e26fcb","subject":"Mirror arista and trilio charms to GitHub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b1fae6833bc2093c491b794915e98291e2e26fcb"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 10:22:35.000000000","tz":0},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2020-07-02 15:25:14.000000000","tz":0},"subject":"Revert \"Create a new repository for Tobiko DevStack plugin\"","message":"Revert \"Create a new repository for Tobiko DevStack plugin\"\n\nThis reverts commit c7ca7592a98b9b9e229f3e73521f5b12a4945a2e.\n\nThe project has been created with a mispelled name.\n\nChange-Id: I77b91dd03a28f8960d7294c1ef086f4b46b02ccd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/361e40a8a5301627821108add101f69ff4564e93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/361e40a8a5301627821108add101f69ff4564e93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
