)]}'
{"id":"openstack%2Fproject-config~739605","triplet_id":"openstack%2Fproject-config~master~Ia6ced2330df3f6ee4d2794743ef9436aac800b96","project":"openstack/project-config","branch":"master","topic":"674091","hashtags":[],"change_id":"Ia6ced2330df3f6ee4d2794743ef9436aac800b96","subject":"Air Handler is failed - turn OE off for now","status":"MERGED","created":"2020-07-06 21:15:19.000000000","updated":"2020-07-06 21:48:22.000000000","submitted":"2020-07-06 21:35:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739605-1594071321667-4780df3a","meta_rev_id":"127ccc8670c234d58060135cae6525c52d193440","_number":739605,"virtual_id_number":739605,"owner":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 21:35:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2020-07-06 21:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-06 21:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-06 21:19:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 21:15:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-07-06 21:19:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-07-06 21:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ee9ce8528ced57c7773fe992916d40c214f97aa8","author":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"date":"2020-07-06 21:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"207e290137aeb84c8933f0855dc3baa1e96aecc0","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-06 21:15:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a7518b6c54ff4ab768e9f07d7797b688629ca589","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-07-06 21:19:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"167369654fb86ccecbe731aab6caffbdf9229931","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 21:26:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e182b711b46b493088a97fc39344b4ca : SUCCESS in 9m 49s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/68a20fccf0d04328b0075f5eb00ffb77 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"20c173f51f6f456e0a7fbb9330fd6bee514fc9f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 21:26:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06a2f9eb57002233730b767a50858b240ed148a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 21:35:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cea92b34361b4f04b2a9d376b1c7a72e : SUCCESS in 8m 04s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/15237b53c8b24bd8834bff5f79bee07a : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9f5b2a2110ab2910596629e70845cfcbe421d74f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 21:35:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"10b64acbca18675d9c5457b7c900fa6dff86ced5","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 21:48:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/7173fe92296c4b8e8fa59a925135c758 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"da3f5808039fc7ee559ad542abf406323d3745e1","revisions":{"da3f5808039fc7ee559ad542abf406323d3745e1":{"kind":"REWORK","_number":1,"created":"2020-07-06 21:15:19.000000000","uploader":{"_account_id":20754,"name":"Donny Davis","email":"donny@fortnebula.com","username":"donnydavis"},"ref":"refs/changes/05/739605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/05/739605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/739605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/739605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/05/739605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/05/739605/1"}}},"commit":{"parents":[{"commit":"e0545f74e885686dfda8f064f3d9e6cf6465ebb6","subject":"Add node-labeler to the vexxhost tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0545f74e885686dfda8f064f3d9e6cf6465ebb6"}]}],"author":{"name":"donnydavis","email":"donny@fortnebula.com","date":"2020-07-06 21:15:12.000000000","tz":-240},"committer":{"name":"donnydavis","email":"donny@fortnebula.com","date":"2020-07-06 21:15:12.000000000","tz":-240},"subject":"Air Handler is failed - turn OE off for now","message":"Air Handler is failed - turn OE off for now\n\nChange-Id: Ia6ced2330df3f6ee4d2794743ef9436aac800b96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da3f5808039fc7ee559ad542abf406323d3745e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da3f5808039fc7ee559ad542abf406323d3745e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
