)]}'
{"id":"openstack%2Fproject-config~741247","triplet_id":"openstack%2Fproject-config~master~I12ca39af9d98bc20e44b71845b68b7a59c4c7dcf","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I12ca39af9d98bc20e44b71845b68b7a59c4c7dcf","subject":"added 2 vexxhost repos","status":"MERGED","created":"2020-07-15 15:27:00.000000000","updated":"2020-07-15 18:03:28.000000000","submitted":"2020-07-15 17:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741247-1594834947466-bdb66bb5","meta_rev_id":"b51355caef0a15b50bc6f10dc080fb847f406d75","_number":741247,"virtual_id_number":741247,"owner":{"_account_id":31752,"name":"Vladimir Blando","email":"vblando@vexxhost.com","username":"vblando"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-15 17:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2020-07-15 16:13:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":1,"date":"2020-07-15 16:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-15 16:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},{"value":1,"date":"2020-07-15 17:28:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 16:13:55.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2020-07-15 16:14:13.000000000","updated_by":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"reviewer":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"state":"REVIEWER"},{"updated":"2020-07-15 16:15:53.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-07-15 17:28:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-07-15 17:42:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac1174c0bac6e7ad5d5396dcd05694cb71451618","author":{"_account_id":31752,"name":"Vladimir Blando","email":"vblando@vexxhost.com","username":"vblando"},"date":"2020-07-15 15:27:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"810a849c51a35081e8f318cc22c5d84faeab47ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 15:37:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/24675fbbd4c94d48a7f2f8eb2a8ca68b : SUCCESS in 9m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a1b8f99ad6e042708f008422a84e43e6 : FAILURE in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"531ed452a1812aa5d22bf0d547549f738a6f5a7c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-15 15:39:32.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc8e1f68210894a3e5044f6ce27064c4f3cbb656","author":{"_account_id":31752,"name":"Vladimir Blando","email":"vblando@vexxhost.com","username":"vblando"},"date":"2020-07-15 15:41:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cfe95a0b38ed1c48f8915dfd0fd3491be24932a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 15:55:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6ba0198282864b5e8a64a022bea2c9dd : SUCCESS in 8m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e0d0d570820a417aabf7977a77d9fd68 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"3561d5bfd0d2508c2af6169557ea7b7872fafaa8","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2020-07-15 16:13:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a45848ae9fa5a869ee9dc56527bf0be5f9a6aa9a","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-07-15 16:14:13.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"51c1e3c6d9564911fdc3d3526c06571e25cbd135","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-15 16:15:18.000000000","message":"Patch Set 2: Workflow-1\n\nWIP so that this can be used to verify our gitea upgrade which is happening shortly.","accounts_in_message":[],"_revision_number":2},{"id":"3ef28e44023ea3b74fa87940f136d46f019e77db","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-15 16:15:53.000000000","message":"Patch Set 2: Code-Review+2\n\nok from my side!","accounts_in_message":[],"_revision_number":2},{"id":"d55ea2da2ba787554fa1ef57297f9bc298e43f60","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-15 17:28:36.000000000","message":"Patch Set 2: Workflow+1\n\nGitea upgrade is complete and seems happy enough.","accounts_in_message":[],"_revision_number":2},{"id":"0f428f9490be5326ceedfbf20f2a7388bfe25c0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:28:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e476a1892bdabf6ed54ccf93f9bf5f06adad58a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:42:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e3e71c061ff0464bbb21b21009702624 : SUCCESS in 11m 48s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e9cdd12b1b7d484fa10d85ef66bdd887 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"570ad24f1154a6751c0ce6565d5e9eccada74de7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 17:42:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b51355caef0a15b50bc6f10dc080fb847f406d75","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 18:03:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/571118955be14ef88e62ad0161366195 : SUCCESS in 3m 54s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/752012bf6edf4a9fb1718334b746b18a : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0a213f3a7dbb72e5d791d7b0dc9277eebc02fff1","revisions":{"308b3024e5beddb11e71bd5fcc4afa80d308a88e":{"kind":"REWORK","_number":1,"created":"2020-07-15 15:27:00.000000000","uploader":{"_account_id":31752,"name":"Vladimir Blando","email":"vblando@vexxhost.com","username":"vblando"},"ref":"refs/changes/47/741247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/741247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/741247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/741247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/741247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/741247/1"}}},"commit":{"parents":[{"commit":"8c71a28df67a305c05b8c034f08f085d28d5ff24","subject":"More helpful description for openstack/openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c71a28df67a305c05b8c034f08f085d28d5ff24"}]}],"author":{"name":"Vladimir Blando","email":"vblando@vexxhost.com","date":"2020-07-15 15:26:24.000000000","tz":480},"committer":{"name":"Vladimir Blando","email":"vblando@vexxhost.com","date":"2020-07-15 15:26:24.000000000","tz":480},"subject":"added 2 vexxhost repos","message":"added 2 vexxhost repos\n\nChange-Id: I12ca39af9d98bc20e44b71845b68b7a59c4c7dcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/308b3024e5beddb11e71bd5fcc4afa80d308a88e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/308b3024e5beddb11e71bd5fcc4afa80d308a88e"}]},"branch":"refs/heads/master"},"0a213f3a7dbb72e5d791d7b0dc9277eebc02fff1":{"kind":"REWORK","_number":2,"created":"2020-07-15 15:41:40.000000000","uploader":{"_account_id":31752,"name":"Vladimir Blando","email":"vblando@vexxhost.com","username":"vblando"},"ref":"refs/changes/47/741247/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/47/741247/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/741247/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/741247/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/47/741247/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/47/741247/2"}}},"commit":{"parents":[{"commit":"8c71a28df67a305c05b8c034f08f085d28d5ff24","subject":"More helpful description for openstack/openstack","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c71a28df67a305c05b8c034f08f085d28d5ff24"}]}],"author":{"name":"Vladimir Blando","email":"vblando@vexxhost.com","date":"2020-07-15 15:26:24.000000000","tz":480},"committer":{"name":"Vladimir Blando","email":"vblando@vexxhost.com","date":"2020-07-15 15:41:01.000000000","tz":480},"subject":"added 2 vexxhost repos","message":"added 2 vexxhost repos\n\nChange-Id: I12ca39af9d98bc20e44b71845b68b7a59c4c7dcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a213f3a7dbb72e5d791d7b0dc9277eebc02fff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0a213f3a7dbb72e5d791d7b0dc9277eebc02fff1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
