)]}'
{"id":"openstack%2Fproject-config~742337","triplet_id":"openstack%2Fproject-config~master~I25fb4a3b1237b4027be2dede58d743212a6b4dda","project":"openstack/project-config","branch":"master","topic":"pypa-cryptography","hashtags":[],"change_id":"I25fb4a3b1237b4027be2dede58d743212a6b4dda","subject":"Add repo for pyca/cryptography Zuul jobs","status":"ABANDONED","created":"2020-07-22 03:36:34.000000000","updated":"2020-07-28 20:37:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c057c130f1394207a8ea3bde7cf275681ac8ad5","_number":742337,"virtual_id_number":742337,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-22 04:00:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-07-22 04:33:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-22 04:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0351f1a873c2e0670ec29e7a64586263597eb604","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-22 03:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de64f2288d4581261435a38f7d6a76f256fa8c7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 03:45:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63e250d0c2964acd9076035741089e34 : SUCCESS in 7m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fbf528a6eca749c597570def0f1bf94e : FAILURE in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"76d40e5a2ac515d5f59dbe68a7c7f993d183dc0c","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-22 03:53:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7108862f19ac1b269bd138eb00a2e8f46f30642b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 04:00:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6a72b439be4c4d7db36f89370de1edc5 : SUCCESS in 6m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8e494b4ea795480581a5e07f24904dd5 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"0b2bbf52435858b4ae13f43f23d27a96b1305645","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-22 04:33:27.000000000","message":"Patch Set 2: Workflow-1\n\nLet\u0027s make sure everyone is on board with this","accounts_in_message":[],"_revision_number":2},{"id":"e15706172436969cada724215abd0a044df323d2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-28 20:37:57.000000000","message":"Abandoned\n\nwe\u0027re taking a separate tenant approach","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7394ca0c8e8196be70d656554aa08fbc0299a04c","revisions":{"a1ba04940643559b64d3e6e7686375814c987a25":{"kind":"REWORK","_number":1,"created":"2020-07-22 03:36:34.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/37/742337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/742337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/742337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/742337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/742337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/742337/1"}}},"commit":{"parents":[{"commit":"715fbc4992a5e51ec98f4519b610d46f2332a306","subject":"Add pyca/cryptography to zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/715fbc4992a5e51ec98f4519b610d46f2332a306"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-07-22 03:32:43.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-07-22 03:32:43.000000000","tz":600},"subject":"Add repo for pyca/cryptography Zuul jobs","message":"Add repo for pyca/cryptography Zuul jobs\n\nBecause we aren\u0027t gating on the upstream project, we can\u0027t read the\nZuul config from there because it might break Zuul configuration.\n\nCreate a new project to hold the jobs/playbooks/roles for running\ntesting againts this repo.\n\nIt seems like something we might want to do more for other projects\ntoo.  That\u0027s why I put this under a \"zuul-config-*\" namespace; I feel\nlike it might be a generically good place to keep things.\n\nChange-Id: I25fb4a3b1237b4027be2dede58d743212a6b4dda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1ba04940643559b64d3e6e7686375814c987a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a1ba04940643559b64d3e6e7686375814c987a25"}]},"branch":"refs/heads/master"},"7394ca0c8e8196be70d656554aa08fbc0299a04c":{"kind":"REWORK","_number":2,"created":"2020-07-22 03:53:16.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/37/742337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/37/742337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/742337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/742337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/37/742337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/37/742337/2"}}},"commit":{"parents":[{"commit":"715fbc4992a5e51ec98f4519b610d46f2332a306","subject":"Add pyca/cryptography to zuul config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/715fbc4992a5e51ec98f4519b610d46f2332a306"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-07-22 03:32:43.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-07-22 03:52:55.000000000","tz":600},"subject":"Add repo for pyca/cryptography Zuul jobs","message":"Add repo for pyca/cryptography Zuul jobs\n\nBecause we aren\u0027t gating on the upstream project, we can\u0027t read the\nZuul config from there because it might break Zuul configuration.\n\nCreate a new project to hold the jobs/playbooks/roles for running\ntesting againts this repo.\n\nIt seems like something we might want to do more for other projects\ntoo.  That\u0027s why I put this under a \"zuul-config-*\" namespace; I feel\nlike it might be a generically good place to keep things.\n\nChange-Id: I25fb4a3b1237b4027be2dede58d743212a6b4dda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7394ca0c8e8196be70d656554aa08fbc0299a04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7394ca0c8e8196be70d656554aa08fbc0299a04c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
