)]}'
{"id":"openstack%2Fproject-config~743217","triplet_id":"openstack%2Fproject-config~master~Id48fff64d13c258d9f22908debfad86c5f089bf5","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Id48fff64d13c258d9f22908debfad86c5f089bf5","subject":"Add nested-virt-ubuntu-focal label","status":"MERGED","created":"2020-07-27 12:55:37.000000000","updated":"2020-09-02 23:15:41.000000000","submitted":"2020-09-02 23:06:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743217-1599088016818-436c11b4","meta_rev_id":"ad39a69e7437968516f8ef73366c65701362f31b","_number":743217,"virtual_id_number":743217,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-02 23:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-02 22:56:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-07-27 13:23:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-28 20:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-09-02 22:56:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-27 13:23:45.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-07-28 20:06:01.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-09-02 22:56:54.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-02 23:06:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12f4b2d252e8aef0aad97b3ca54ee5d59c2a8e44","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-07-27 12:55:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bdbd50950d196b31f0758e1be597ba6163bb33e","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-07-27 13:02:58.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8b6fdb906b11a0437bb5a4e2e98694038b74c93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 13:18:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9971407da86f475d996692833746370f : SUCCESS in 8m 07s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/b945a20a9ed0484e9e55cda29720dc5f : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"d85b94bb5b3bc2ca280c86e67668339ee680e449","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-07-27 13:23:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d945f0955f8b655b3525f3d785eecd3a47be6de9","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-07-28 20:06:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35a4a9b3ef6b16f5bea9010726a8a7d0659668ed","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-02 22:56:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7c5675f46bf80b69744519f4bf37e6201432bd1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 22:57:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"29c1051e7956d0816daabb30b106c46ad6624669","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 23:06:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8172e3cae9c4c91a8ee000f5153254a : SUCCESS in 7m 25s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/344bd54f488f4a02ac85cfdb6e29bcf5 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"725788457e13ab75904fb0747e861689afeb2f5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 23:06:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ad39a69e7437968516f8ef73366c65701362f31b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 23:15:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/4b68a23d3b5e41239b7e9f26e473abdf : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"357e2ef12eb67eed5a2dc2f7ad8e5c2222eae6d5","revisions":{"b3a9df0a8dcf4f429e7337a055d5bc29c91b2eeb":{"kind":"REWORK","_number":1,"created":"2020-07-27 12:55:37.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/17/743217/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/743217/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/743217/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/743217/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/743217/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/743217/1"}}},"commit":{"parents":[{"commit":"9492fdef299bfa04e0c00871b558dc157f4ef5bf","subject":"Merge \"Update ceph grafana for nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9492fdef299bfa04e0c00871b558dc157f4ef5bf"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-07-27 12:49:04.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-07-27 12:49:04.000000000","tz":-240},"subject":"Add nested-virt-ubuntu-focal label","message":"Add nested-virt-ubuntu-focal label\n\nUbuntu Focal has a newer libvirt version than Bionic (4.0.0 vs 6.0.0).\nBy adding a Focal-flavored nested-virt label, features made possible\nby a more recent libvirt version can be tested in the gate.\nSpecifically, whitebox-tempest-plugin tests Nova\u0027s hw_video_type image\nproperty. Support for the \u0027none\u0027 value was added in libvirt 4.6.0.\n\nChange-Id: Id48fff64d13c258d9f22908debfad86c5f089bf5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3a9df0a8dcf4f429e7337a055d5bc29c91b2eeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3a9df0a8dcf4f429e7337a055d5bc29c91b2eeb"}]},"branch":"refs/heads/master"},"357e2ef12eb67eed5a2dc2f7ad8e5c2222eae6d5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-27 13:02:58.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/17/743217/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/743217/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/743217/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/743217/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/743217/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/743217/2"}}},"commit":{"parents":[{"commit":"9492fdef299bfa04e0c00871b558dc157f4ef5bf","subject":"Merge \"Update ceph grafana for nova\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9492fdef299bfa04e0c00871b558dc157f4ef5bf"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-07-27 12:49:04.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2020-07-27 13:02:45.000000000","tz":-240},"subject":"Add nested-virt-ubuntu-focal label","message":"Add nested-virt-ubuntu-focal label\n\nUbuntu Focal has a newer libvirt version than Bionic (4.0.0 vs 6.0.0).\nBy adding a Focal-flavored nested-virt label, features made possible\nby a more recent libvirt version can be tested in the gate.\nSpecifically, whitebox-tempest-plugin tests Nova\u0027s hw_video_type image\nproperty. Support for the \u0027none\u0027 value was added in libvirt 4.6.0.\n\nChange-Id: Id48fff64d13c258d9f22908debfad86c5f089bf5\nNeeded-by: https://review.opendev.org/#/c/742014/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/357e2ef12eb67eed5a2dc2f7ad8e5c2222eae6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/357e2ef12eb67eed5a2dc2f7ad8e5c2222eae6d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
