)]}'
{"id":"openstack%2Fproject-config~743766","triplet_id":"openstack%2Fproject-config~master~Iad535a9b963a3308d11d314484eb2c5d11e2008f","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"Iad535a9b963a3308d11d314484eb2c5d11e2008f","subject":"Add Keystone Kerberos charm to OpenStack charms","status":"MERGED","created":"2020-07-29 13:57:11.000000000","updated":"2020-07-30 16:57:16.000000000","submitted":"2020-07-30 16:38:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743766-1596127114955-e2009d81","meta_rev_id":"5d8b396d8c28d94defd6cd0b2c4f956eda6e08c8","_number":743766,"virtual_id_number":743766,"owner":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-30 16:38:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-07-30 16:27:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-07-30 07:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-30 13:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-07-30 16:27:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 07:28:45.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2020-07-30 13:22:42.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-07-30 16:27:21.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-07-30 16:38:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"760b025281fe1d88ee9e7d8920de43555c482b45","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2020-07-29 13:57:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a545dce7d9ffe096f1219eb8358ec0ea8f57c74a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:08:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1d388468d63a4f1fa8de2c5a21bae03d : SUCCESS in 7m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ca7d9b035cb64fa8ab7335c30f2d56b2 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e4eeb43f308e544ee4134d02ec293e5bb08630a8","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2020-07-29 14:10:38.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ba0ed8302730e94c6c5e5896055580349d4b08a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 14:26:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9bc3b86755134ec0959ea80c2da832e4 : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8dd0e48d369845e099d894fd14d7f5e2 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"efdeefd85eadcea3919e6846b3495b07abbb5463","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2020-07-30 07:28:45.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"db04be9bbbf9f6dc4d864c5c77d1b43a6505433b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-07-30 13:22:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ea4257a743758c3318c7b2578e7f048be52afc69","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-30 16:27:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"07d77624d5c18da9220b7de1a79deb85268c42bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:27:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d502d36a37b52525ec20420d004b5b1bb8dfe6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:38:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7cb009919d894fe0859e8f6f983a628a : SUCCESS in 6m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/294310b1acd54c04bed45488449a7cc7 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"460cd5e80c24054aae6f798f83c3df2ebef7ae63","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:38:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"5ea18a5cf8346cb0f2f6c3c48815a3c3d7d7c6d8","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 16:57:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/d61d4e37c1c7435dae7730ec2fc52bd1 : SUCCESS in 2m 23s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/cff74ca85035469dbf1f5c2644ba6bc4 : SUCCESS in 10m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","revisions":{"a7046e8b915de4eee02098a04c162eccef625241":{"kind":"REWORK","_number":1,"created":"2020-07-29 13:57:11.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/66/743766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/743766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/743766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/743766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/743766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/743766/1"}}},"commit":{"parents":[{"commit":"2604554c4f7af70f8513d3d811bec12b4ce63f1b","subject":"Merge \"Add pyca 3rd-party CI tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2604554c4f7af70f8513d3d811bec12b4ce63f1b"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-07-29 13:56:36.000000000","tz":120},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-07-29 13:56:36.000000000","tz":120},"subject":"Add Keystone Kerberos charm to OpenStack charms","message":"Add Keystone Kerberos charm to OpenStack charms\n\nAdd new charm that provides Keystone the ability to authenticate with a\nKerberos server.\n\nChange-Id: Iad535a9b963a3308d11d314484eb2c5d11e2008f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7046e8b915de4eee02098a04c162eccef625241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a7046e8b915de4eee02098a04c162eccef625241"}]},"branch":"refs/heads/master"},"2455578066f6c4611fc7efc36c78e73c1b4ba4c3":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-29 14:10:38.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/66/743766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/743766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/743766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/743766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/743766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/743766/2"}}},"commit":{"parents":[{"commit":"2604554c4f7af70f8513d3d811bec12b4ce63f1b","subject":"Merge \"Add pyca 3rd-party CI tenant\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2604554c4f7af70f8513d3d811bec12b4ce63f1b"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-07-29 13:56:36.000000000","tz":120},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-07-29 14:10:24.000000000","tz":120},"subject":"Add Keystone Kerberos charm to OpenStack charms","message":"Add Keystone Kerberos charm to OpenStack charms\n\nAdd new charm that provides Keystone the ability to authenticate with a\nKerberos server.\n\nChange-Id: Iad535a9b963a3308d11d314484eb2c5d11e2008f\nRequired-By: I0a3d96268157b056685e954ee8bd2103a32e6e63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
