)]}'
{"id":"openstack%2Fproject-config~744110","triplet_id":"openstack%2Fproject-config~master~I9809fdf1745ac8573502023c63f0f20e98862af1","project":"openstack/project-config","branch":"master","topic":"complete-retirement-networking-l2gw","hashtags":[],"change_id":"I9809fdf1745ac8573502023c63f0f20e98862af1","subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","status":"MERGED","created":"2020-07-31 09:53:23.000000000","updated":"2020-08-26 22:30:51.000000000","submitted":"2020-08-26 22:05:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744110-1598479523780-8a9bf4a0","meta_rev_id":"55d3fc395fb77307959ced863ec887c4d9e2e855","_number":744110,"virtual_id_number":744110,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-26 22:05:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-26 09:52:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":1,"date":"2020-08-26 17:58:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-08-26 10:41:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-26 15:03:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-08-26 21:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":1,"date":"2020-08-26 21:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 09:52:38.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-26 10:41:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-26 15:03:23.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-08-26 17:58:22.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-08-26 21:56:21.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-08-26 22:05:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45511efdb1fca2b4fd835a2bf2b0ae54e96f7b6a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-31 09:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4878070b1edf0b013f756bc18b848afe2ccb123c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 10:04:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7eaf5264aa6243028d1a85a9241a8adc : SUCCESS in 6m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2eddadcc942b44ccaf200ff129fa0a32 : FAILURE in 3m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c3c0291d1d0d1529c6f5f796fdc064d5be531d0d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-07-31 11:03:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33d63226d32718435538430e17f32504d40f7fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 11:20:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2098ab64781f4135ac50c40fdd21740d : SUCCESS in 7m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/671b4f2efa804bffa46ec4a334c23870 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"103364ad8950fa87083a49c20c50eae76ee93b4d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-05 16:40:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nthanks, let\u0027s separate the acl file also form openstack/ folder","accounts_in_message":[],"_revision_number":2},{"id":"5c9187e092d5ae78deb0daf9f6e963790ce9ff87","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-06 06:35:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0522e5f578f6a408e1e99c5c66405ffdbdbc39c4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-06 06:36:44.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThanks gmann, I just reached again to this in my todo list :-)","accounts_in_message":[],"_revision_number":3},{"id":"5a4a940dd107175ebc53033adf5af9714e336d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 06:52:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6add9d87ad9d4d1fa27d7a7f2572100e : SUCCESS in 7m 44s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1cc5898e1e01484dad176daf4b3c3143 : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"bce0da91315e36f87a5524adc97e856179b0ce8a","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 13:16:36.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nThis change misses to add the repos to zuul/main.yaml","accounts_in_message":[],"_revision_number":3},{"id":"f4255b0b758954874cec1e2a16f21ed6b38145b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-06 13:23:46.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dfb5a631ebb03836db09d954ad332ad8784d2022","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 13:26:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7ae7b50d1fcf5db1a0ff544c5a08c13cfb8326a5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-06 13:31:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"185d06bbfd38dd8e88ac8f2965dd64dfaa4e1f72","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 13:43:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e8eec961422168f3a757b10100e2b41b648502c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-06 13:59:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"caf613a437b094de7bbd5138beeda632d5ea0e06","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 19:31:00.000000000","message":"Patch Set 3:\n\nAs discussed on #openstack-infra: I\u0027d like to see a plan - Gmann proposed to write it in https://etherpad.opendev.org/p/tc-retirement-cleanup - for what you do with the openstack and x repos and their stable branches.\n\nAlso, I\u0027d like to see all patches for this work up so that we can review them together. Some patches can be WIPed if needed or have a Depends-On.","accounts_in_message":[],"_revision_number":3},{"id":"6a723f499865f7c64396307612cd844b3efaff10","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-24 13:34:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60ea4ee26bb629e79bc061669306a41184e06269","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-24 14:04:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"194f9dfadb15bd5efde1dbed92e1e2fdeb4eda46","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-24 14:05:30.000000000","message":"Patch Set 3:\n\nCould you update the change to add the new repo to zuul/main.yaml.","accounts_in_message":[],"_revision_number":3},{"id":"3f368185555757445fbf34790c4be3ba4a1e0987","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-08-24 15:35:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"816e0eb3497865ffde6fb1c4209097158c154edc","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-24 15:41:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a49b1778aec006dec383488634e28c2f58013ec3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-25 16:32:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4e455b88e4de42a2a7379105768e6a025c39fd00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 16:41:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f48cd4f1659a4609a27594ebd75824b8 : SUCCESS in 5m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/df836b5d2dfa42dcb8860171668f7fe2 : FAILURE in 4m 24s","accounts_in_message":[],"_revision_number":4},{"id":"722a836af75efb88d70378897390ce93ad03a315","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-25 19:02:19.000000000","message":"Patch Set 4: Code-Review-1\n\nNeeds change for zuul/main.yaml","accounts_in_message":[],"_revision_number":4},{"id":"9902a41069e683390933ff27ff146578a26d0551","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-26 07:50:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"68be2748a56afd05f308bb88b418b90ef3ffe51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:02:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e40c301d29a0434f8fa2965e2c5396aa : SUCCESS in 7m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/efa841cfb59f4f1795c51c572cbaa305 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":5},{"id":"53ce20806a05d24e99e38d279523a4ec2b530cde","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-26 08:20:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6bfebe4a6bfec037ba887f26bf453973cfe22467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:42:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3574c62d353b44b7812134c45abcc922 : SUCCESS in 7m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6ac05b0cca8a4715859a67f444416cf0 : FAILURE in 5m 19s","accounts_in_message":[],"_revision_number":6},{"id":"0593e79a40986d87175dfd741725c67f2bd068b4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-26 08:46:38.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5007c4da4de897337324862cf3b6c622cfe6808e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-26 08:48:29.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"ca4aa89187cda5b84f64cdbaff77657f9f82fcf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 08:58:14.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cc8ef9d71e90469dac22467fcb714300 : SUCCESS in 7m 05s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/071399e4e50d458192a66b47b1fd457b : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":7},{"id":"8687aa15c8c9588d21733913e81fa67786d47724","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-26 09:05:25.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"71216bd73566050f6b7b607867c086f58e59a246","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2020-08-26 09:35:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"26853d75c43bd8fe92ed89a1a18bd93f934a69de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 09:45:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a4f603b4bb51472bac0cb7e6b96b76c4 : SUCCESS in 5m 31s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/bfb75de9863545a4bb20d60404eee5b8 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":8},{"id":"86d1842aa8211dfad2db8669a00578b930cdafd2","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-26 09:52:38.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"bb48f1409a2ebf9dcf24d54818e8ffd62d69cd03","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-26 10:41:22.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"0fa8e432d7cbe56ae562ef7ea76818a5693c7dc0","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-26 15:03:23.000000000","message":"Patch Set 8: Code-Review+1\n\nthanks Lajos","accounts_in_message":[],"_revision_number":8},{"id":"5e2df7ba2bad5550147da1a27ebbbaf13e91fece","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-08-26 17:58:22.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"507c16938e0968581e10937b63ef7669f4b5f593","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-26 21:56:21.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ece91054c0567ba6dcad8572a5e3286679b9a3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 21:56:36.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"dac0771ee4786060c819debda801c667b24d61a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:05:23.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4e7f2223dff748648601fa77c06473ba : SUCCESS in 6m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b36c66e06b374ee1b713e3d4767f814b : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":8},{"id":"dd6f6ff1219acecdc8bffdae3f306df161182e25","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:05:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"f7e13f154db6cb39e7f04feb8d2b9b68f38b6ae7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 22:30:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a109ff22ff38492c8185ac2d9f544181 : SUCCESS in 2m 47s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/0af57623a83b42f08b5404dd2b65b2f5 : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6f6e4da925d3fb856865ffde93db4611aef212e6","revisions":{"edc2cf6312ff976ee4a41370bfd204742bfd0400":{"kind":"REWORK","_number":1,"created":"2020-07-31 09:53:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/1"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edc2cf6312ff976ee4a41370bfd204742bfd0400"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edc2cf6312ff976ee4a41370bfd204742bfd0400"}]},"branch":"refs/heads/master"},"8dbf82c98d6b2eba7d89b6b14433fafcfa048a24":{"kind":"REWORK","_number":2,"created":"2020-07-31 11:03:20.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/2"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 11:02:26.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8dbf82c98d6b2eba7d89b6b14433fafcfa048a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8dbf82c98d6b2eba7d89b6b14433fafcfa048a24"}]},"branch":"refs/heads/master"},"a420306552c79162afb7b7d50b6cb011104aed20":{"kind":"REWORK","_number":3,"created":"2020-08-06 06:35:23.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/3"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-06 06:34:29.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a420306552c79162afb7b7d50b6cb011104aed20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a420306552c79162afb7b7d50b6cb011104aed20"}]},"branch":"refs/heads/master"},"0d113532838e501fb28de8f0179c9266d174001f":{"kind":"REWORK","_number":4,"created":"2020-08-25 16:32:03.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/4"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-25 16:31:05.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0d113532838e501fb28de8f0179c9266d174001f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0d113532838e501fb28de8f0179c9266d174001f"}]},"branch":"refs/heads/master"},"e801533ebac5afd7f4eb3fb1367e43b1cb934205":{"kind":"REWORK","_number":5,"created":"2020-08-26 07:50:10.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/5"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-26 07:49:17.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e801533ebac5afd7f4eb3fb1367e43b1cb934205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e801533ebac5afd7f4eb3fb1367e43b1cb934205"}]},"branch":"refs/heads/master"},"1504d36edab0de5565e962007c40ef08b71bc21a":{"kind":"REWORK","_number":6,"created":"2020-08-26 08:20:13.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/6"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-26 08:19:12.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1504d36edab0de5565e962007c40ef08b71bc21a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1504d36edab0de5565e962007c40ef08b71bc21a"}]},"branch":"refs/heads/master"},"ebbdde458f395f58a6d66e78b0d6fd6057d43710":{"kind":"REWORK","_number":7,"created":"2020-08-26 08:48:29.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/7"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"Lajos Katona","email":"katonalala@gmail.com","date":"2020-08-26 08:48:26.000000000","tz":0},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebbdde458f395f58a6d66e78b0d6fd6057d43710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ebbdde458f395f58a6d66e78b0d6fd6057d43710"}]},"branch":"refs/heads/master"},"6f6e4da925d3fb856865ffde93db4611aef212e6":{"kind":"REWORK","_number":8,"created":"2020-08-26 09:35:20.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/10/744110/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/10/744110/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/10/744110/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/10/744110/8"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-07-31 09:51:10.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2020-08-26 09:34:31.000000000","tz":120},"subject":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/","message":"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\n\nnetworking-l2gw is not a neutron stadium project, and not governed by\nopenstack, so it is time to move it to x/ namespace.\n\nChange-Id: I9809fdf1745ac8573502023c63f0f20e98862af1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f6e4da925d3fb856865ffde93db4611aef212e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6f6e4da925d3fb856865ffde93db4611aef212e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
