)]}'
{"id":"openstack%2Fproject-config~744195","triplet_id":"openstack%2Fproject-config~master~Ie29a071b1b8916e0d09990b97bf589bf2b0fc50a","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"Ie29a071b1b8916e0d09990b97bf589bf2b0fc50a","subject":"Re-add vexxhost-specific labels","status":"MERGED","created":"2020-07-31 15:11:13.000000000","updated":"2020-07-31 16:00:29.000000000","submitted":"2020-07-31 15:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744195-1596210833221-60d190a1","meta_rev_id":"54d0725441aad9426d1e5164823ff0435fcd7c4b","_number":744195,"virtual_id_number":744195,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 15:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-07-31 15:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-07-31 15:18:17.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-31 15:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-07-31 15:18:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 15:18:17.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-07-31 15:18:57.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-07-31 15:25:14.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-07-31 15:53:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"83a30c18781a40208b2360ef6334aa8112ca5ee9","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-31 15:11:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0d57e9f7265f7d59f72959f0afb125cdca49243","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-07-31 15:13:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"648c1413c28c59696b70735ad84f9cdceebf44ee","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-07-31 15:13:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b631376fe827588f661191f2797d39718cfa4e9c","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-31 15:14:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25dd34f6dd45ab0f231bbc582718373b03c54eee","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-31 15:15:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a8fd399be63a0972ce2208c8cc285d178c283d3","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-07-31 15:17:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f0716a0775f4451da0021780fd9ceeb0757f7eb","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-07-31 15:18:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b0d4780c9489ce145b876d795d138c55edd8b149","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-07-31 15:18:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c1702070e262fe12f79cc2c77394941ebd2cf983","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-07-31 15:25:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2ecf4ffad5505bdb2cbbf04355e5faba53e02707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:38:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/72cbb0b3788a496b9aec6ef77d3ed1dd : SUCCESS in 9m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/80b8e06b0d3348c4922a1f65c604be3d : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"72ee68fd7aed3329bc0d73476acad5b768dbde96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:38:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b70bb7d3d4d11806d90d857db873ac73af9d94e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:53:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cde3fa9977a64f0e8f0147398654858f : SUCCESS in 7m 01s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/4ef9097a877042669b88d25c095cfc88 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":2},{"id":"e75b6f9f0122ac1aebd8052f76e4e55654bc32f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 15:53:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"54d0725441aad9426d1e5164823ff0435fcd7c4b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 16:00:29.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/2d219c447ad645019573222691e0f9e5 : FAILURE in 6m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"34f1a5543a20c50d40d4e82d4beb24e0b7f5e3eb","revisions":{"5eb9db88b3e8cb74804b54059c1dccfcb0543bff":{"kind":"REWORK","_number":1,"created":"2020-07-31 15:11:13.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/95/744195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/744195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/744195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/744195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/744195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/744195/1"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-31 15:09:58.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-31 15:09:58.000000000","tz":-240},"subject":"Re-add vexxhost-specific labels","message":"Re-add vexxhost-specific labels\n\nThese labels were not migrated when we moved to running inside our new\nhardware inside the Montreal datacenter.\n\nThose largely should disappear as soon as we move our operator\ndevelopment to be multi-node to avoid having a \u0027unicorn\u0027 flavor.  We\u0027d\nlike to use them temporarily to unblock our progress for now.\n\nChange-Id: Ie29a071b1b8916e0d09990b97bf589bf2b0fc50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5eb9db88b3e8cb74804b54059c1dccfcb0543bff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5eb9db88b3e8cb74804b54059c1dccfcb0543bff"}]},"branch":"refs/heads/master"},"34f1a5543a20c50d40d4e82d4beb24e0b7f5e3eb":{"kind":"REWORK","_number":2,"created":"2020-07-31 15:17:15.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/95/744195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/744195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/744195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/744195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/744195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/744195/2"}}},"commit":{"parents":[{"commit":"2455578066f6c4611fc7efc36c78e73c1b4ba4c3","subject":"Add Keystone Kerberos charm to OpenStack charms","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2455578066f6c4611fc7efc36c78e73c1b4ba4c3"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-31 15:09:58.000000000","tz":-240},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2020-07-31 15:16:37.000000000","tz":-240},"subject":"Re-add vexxhost-specific labels","message":"Re-add vexxhost-specific labels\n\nThese labels were not migrated when we moved to running inside our new\nhardware inside the Montreal datacenter.\n\nThose largely should disappear as soon as we move our operator\ndevelopment to be multi-node to avoid having a \u0027unicorn\u0027 flavor.  We\u0027d\nlike to use them temporarily to unblock our progress for now.\n\nChange-Id: Ie29a071b1b8916e0d09990b97bf589bf2b0fc50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/34f1a5543a20c50d40d4e82d4beb24e0b7f5e3eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/34f1a5543a20c50d40d4e82d4beb24e0b7f5e3eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
