)]}'
{"id":"openstack%2Fproject-config~744479","triplet_id":"openstack%2Fproject-config~master~I9e385c4206aa6e9a7315099f10c4b2b1530f3e13","project":"openstack/project-config","branch":"master","topic":"add-ceph-iscsi-and-layers","hashtags":[],"change_id":"I9e385c4206aa6e9a7315099f10c4b2b1530f3e13","subject":"Add Ceph iSCSI charm to OpenStack charms","status":"MERGED","created":"2020-08-03 14:12:32.000000000","updated":"2020-08-04 08:53:37.000000000","submitted":"2020-08-04 08:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744479-1596529946565-8e31e638","meta_rev_id":"cd34303adc2681ac27c6ef18a65613228732c463","_number":744479,"virtual_id_number":744479,"owner":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-04 08:32:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-04 08:25:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-08-04 08:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":2,"date":"2020-08-03 14:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-08-04 08:25:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-03 14:59:23.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-08-04 08:14:10.000000000","updated_by":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"reviewer":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"state":"REVIEWER"},{"updated":"2020-08-04 08:25:58.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-04 08:32:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5896857b37fc7b7114c7770c6d1ca19e1fce02c","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2020-08-03 14:12:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ca30b26e65d226352bb33225fe8e1d3e282ecff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:22:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/6bcf4eecb1b04877b2ab79275e187eae : SUCCESS in 7m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1e3524c191d541d5b87a470e2c72a93a : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cff8a25252950f506ae65e46768891ad1f2acdc1","author":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"date":"2020-08-03 14:46:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2aed5e58eb90ade9e401093b9ef82ba76a88fe86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-03 14:57:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a3fc84fef2504704891a98409cdf3a62 : SUCCESS in 7m 16s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/30c106336fd64359a6aa1097fb0031b9 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"09c50001b83f8c3bbf1b0d86d4d082582a0bc33f","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-03 14:59:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"56028032f1d1a117e66c3ee4d3b9072e7d918f25","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-04 06:22:39.000000000","message":"Patch Set 2: Workflow-1\n\nNeeds PTL+1","accounts_in_message":[],"_revision_number":2},{"id":"f16afac042c8452e75bece57a36f9c22de0469a2","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2020-08-04 08:14:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cf3ee04314f900cf38b1eaf1b4b8b478fff27add","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-04 08:25:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"318430e7afc7a9b60319c9063ba694cbd26129c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 08:26:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"437597df468b0a01f32ce8811df5c2eed5312cc2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 08:32:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4b30bee8b14c9ea7d15da3c9a54fe47ce4791780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 08:32:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/86b2f00af6b94223ad56940c14ab4bd2 : SUCCESS in 5m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e525f4c09da24d88a8a498abb766bdd6 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":2},{"id":"8db0e06b7b7bd5040732ff7881cce34b0dc4f826","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 08:53:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/56090bc655ae45fda49b88af31c08c9f : SUCCESS in 2m 47s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/0a7d375eaacc44298b01b4d04205bf3f : SUCCESS in 9m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a80c942859ce06b99b67afc75d8c7cdc26a38be","revisions":{"6e5f63f1f1ba3a5ccfa943de6833a77a57b587e0":{"kind":"REWORK","_number":1,"created":"2020-08-03 14:12:32.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/79/744479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/744479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/744479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/744479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/744479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/744479/1"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2020-08-03 14:08:35.000000000","tz":60},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2020-08-03 14:11:37.000000000","tz":60},"subject":"Add Ceph iSCSI charm to OpenStack charms","message":"Add Ceph iSCSI charm to OpenStack charms\n\nAdd Ceph iSCSI charm and corresponding interfaces to OpenStack Charms.\n\nChange-Id: I9e385c4206aa6e9a7315099f10c4b2b1530f3e13\nRequired-By: I8d2e1af0b03cc68d161ab0ff6f7f631c513ddca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e5f63f1f1ba3a5ccfa943de6833a77a57b587e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6e5f63f1f1ba3a5ccfa943de6833a77a57b587e0"}]},"branch":"refs/heads/master"},"8a80c942859ce06b99b67afc75d8c7cdc26a38be":{"kind":"REWORK","_number":2,"created":"2020-08-03 14:46:54.000000000","uploader":{"_account_id":12549,"name":"gnuoy","email":"liam.young@canonical.com","username":"gnuoy"},"ref":"refs/changes/79/744479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/79/744479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/744479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/744479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/79/744479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/79/744479/2"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2020-08-03 14:08:35.000000000","tz":60},"committer":{"name":"Liam Young","email":"liam.young@canonical.com","date":"2020-08-03 14:45:04.000000000","tz":60},"subject":"Add Ceph iSCSI charm to OpenStack charms","message":"Add Ceph iSCSI charm to OpenStack charms\n\nAdd Ceph iSCSI charm and corresponding interfaces to OpenStack Charms.\n\nChange-Id: I9e385c4206aa6e9a7315099f10c4b2b1530f3e13\nRequired-By: I8d2e1af0b03cc68d161ab0ff6f7f631c513ddca6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a80c942859ce06b99b67afc75d8c7cdc26a38be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8a80c942859ce06b99b67afc75d8c7cdc26a38be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
