)]}'
{"id":"openstack%2Fproject-config~744592","triplet_id":"openstack%2Fproject-config~master~Ie1b916725c50c3f06a7d8561886da37839864959","project":"openstack/project-config","branch":"master","topic":"new/return_gnocchi","hashtags":[],"change_id":"Ie1b916725c50c3f06a7d8561886da37839864959","subject":"Fork Gnocchi back to openstack as Farfalle","status":"ABANDONED","created":"2020-08-04 02:51:11.000000000","updated":"2021-11-13 19:30:08.000000000","total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97cbf323fb7c6f2e6f601a102885cf18c7c48cc3","_number":744592,"virtual_id_number":744592,"owner":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-03 08:53:07.000000000","_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"date":"2020-08-31 07:21:48.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-28 14:12:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"date":"2020-09-01 01:49:12.000000000","_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":15002,"name":"Bernd Bausch","email":"berndbausch@gmail.com","username":"berndbausch"},{"date":"2020-08-29 07:40:22.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"date":"2020-09-10 11:35:15.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"value":-1,"date":"2020-08-28 14:57:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15002,"name":"Bernd Bausch","email":"berndbausch@gmail.com","username":"berndbausch"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":-1,"date":"2020-08-28 14:44:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"all":[{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":15002,"name":"Bernd Bausch","email":"berndbausch@gmail.com","username":"berndbausch"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"date":"2020-09-03 06:07:53.000000000","_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"date":"2020-08-28 20:45:30.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":-1,"date":"2020-09-04 03:37:44.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":15002,"name":"Bernd Bausch","email":"berndbausch@gmail.com","username":"berndbausch"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-04 02:51:49.000000000","updated_by":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"reviewer":{"_account_id":6484,"name":"Feilong Wang","email":"hustemb@gmail.com","username":"flwang"},"state":"REVIEWER"},{"updated":"2020-08-25 06:48:34.000000000","updated_by":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"reviewer":{"_account_id":6537,"name":"gordon chung","email":"gord@live.ca","username":"chungg"},"state":"REVIEWER"},{"updated":"2020-08-25 23:42:31.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-08-26 13:01:47.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2020-08-26 21:21:21.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-08-26 21:22:26.000000000","updated_by":{"_account_id":15002,"name":"Bernd Bausch","email":"berndbausch@gmail.com","username":"berndbausch"},"reviewer":{"_account_id":15002,"name":"Bernd Bausch","email":"berndbausch@gmail.com","username":"berndbausch"},"state":"REVIEWER"},{"updated":"2020-08-26 21:24:03.000000000","updated_by":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"reviewer":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"state":"REVIEWER"},{"updated":"2020-08-27 06:34:18.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2020-08-27 16:09:37.000000000","updated_by":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"reviewer":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"state":"REVIEWER"},{"updated":"2020-08-27 19:05:56.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-08-28 01:51:29.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-08-28 08:38:57.000000000","updated_by":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"reviewer":{"_account_id":136,"name":"Tim Bell","email":"tim.bell@cern.ch","username":"tim-bell"},"state":"REVIEWER"},{"updated":"2020-08-28 08:48:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-08-28 12:26:52.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-28 13:05:57.000000000","updated_by":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"reviewer":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"state":"REVIEWER"},{"updated":"2020-08-28 14:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-28 14:44:08.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2020-08-28 14:57:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-28 20:45:30.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-08-29 07:40:22.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-08-31 07:21:48.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-09-01 01:49:12.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-09-03 06:07:53.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2020-09-03 08:53:07.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2020-09-10 11:35:15.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"}],"messages":[{"id":"4cb44ce08f117a311574766d51b8d0f8ae275931","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-04 02:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1adf2117fbc300220645c75cc84430c7a9949e6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 03:04:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e7be666aff54abb8b82576b7d3e3ffc : SUCCESS in 7m 17s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5697dd2e316f4d0ab236da66bb327fd3 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":1},{"id":"72625c053855fea79f4c1f9fa7673d0db60b1b7a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-04 09:00:46.000000000","message":"Patch Set 1:\n\nI think we first need to have a discussion on whether to include it in openstack (as part of telemetry team), or just on opendev (under its own namespace).\n\n(I personally think a generic metrics storage DB should be considered distinct from \"openstack\")\n\nI\u0027d suggest starting a ML thread about it.","accounts_in_message":[],"_revision_number":1},{"id":"b893d0dba179a8c726008d0ff5eedf3ecfb53ff5","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-08-04 09:27:43.000000000","message":"Patch Set 1:\n\nWhat makes you think GitHub is currently a block for anyone to contribute?","accounts_in_message":[],"_revision_number":1},{"id":"1221838a4229caa8f58e287d005a380f9abbd922","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-06 01:58:52.000000000","message":"Patch Set 1:\n\n\u003e What makes you think GitHub is currently a block for anyone to\n \u003e contribute?\n\nI can name some of those:\n1. Most of openstack contributors already get used to Gerrit rather than Github for review.\n2. Using Gerrit for project maintainer management.\n3. Using zuul could help the \"new Gnocchi\" project involve together with the whole openstack ecosystem, e.g. using zuul template for testing new python version support, new OS distribution support, etc.","accounts_in_message":[],"_revision_number":1},{"id":"b7423e5e01c2f52320a90146a1036d29e964bb3a","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-08-06 07:16:32.000000000","message":"Patch Set 1:\n\nThanks Lingxian.\n\nCan you send me this list of people that wanted to contribute to Gnocchi and were stopped by it being on GitHub for the reason you mentionned?\n\nI really want to work with these people.","accounts_in_message":[],"_revision_number":1},{"id":"6cf7b1ee45dafb408d92f4da7d7009846d7d90d4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-06 09:58:01.000000000","message":"Patch Set 1:\n\n\u003e Thanks Lingxian.\n \u003e \n \u003e Can you send me this list of people that wanted to contribute to\n \u003e Gnocchi and were stopped by it being on GitHub for the reason you\n \u003e mentionned?\n \u003e \n \u003e I really want to work with these people.\n\nI don\u0027t have a list of names but I am one of those. If you ask the question in openstack-discuss mailing list, I believe you will find the answer by yourself.","accounts_in_message":[],"_revision_number":1},{"id":"e00d31cc93b87e0bf500b3144a93b473535bde6d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-25 06:03:01.000000000","message":"Patch Set 1:\n\n@Julien\n\nDo you see yourself and others who were working on Gnocchi continuing to support it? Because right now we are in a state where OpenStack services rely on it, without any clear visibility on our part that others outside of OpenStack do. Do you have any metrics on non-OpenStack users?\n\nIf we bring it back into our fold it makes it easier to maintain and potentially rally people to work on it, because then the OpenStack community collectively holds governance over it and if the project shows a lack of development effort it makes it easier for someone else to take over.\n\nRight now you and the gnocchixyz group essentially hold governance, which means you are who everyone has to go through for the project, and right now at least from our perspective the project isn\u0027t all that active. You definitely can add interested people to the project/group on github, but you can see why that isn\u0027t exactly the same as OpenStack having governance over the project for when new PTLs or core devs need to be found/added.\n\nThere is nothing stopping the project from being used by external parties, and we can continue to ensure that it can be deployed standalone, but it feels like something that ultimately was built for OpenStack, and is still mostly used for OpenStack, which now rather annoyingly is under external governance.\n\nThis is of course just my perspective on the situation, and Thierry can possibly add a more official voice to the discussion, but I do think it probably will make more sense to bring the project back to OpenStack unless there is a strong drive to still push it for standalone use (which again, we can still maintain).","accounts_in_message":[],"_revision_number":1},{"id":"61e3faab5e5fad489a6979ddab09b87dd8332327","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-08-25 08:11:23.000000000","message":"Patch Set 1:\n\nAdrian, you seem to think GitHub or the Gnocchi maintainer group is the problem for Gnocchi inactivity, so you want to get rid of those.\n\nI\u0027m still waiting for data points that prove that this is the problem that needs to be solved to have contributors to Gnocchi.\n\nFrom where I stand, it\u0027s more likely Gnocchi isn\u0027t active because nobody has an incentive to contribute to it (i.e. they\u0027re not paid to do so, very likely).","accounts_in_message":[],"_revision_number":1},{"id":"7841bce4303d41131fe53b0d77f9b3519a1fc5f5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-25 09:49:10.000000000","message":"Patch Set 1:\n\nLike I said above, this change is not discussing \"returning gnocchi to openstack\", it\u0027s just discussing returning Gnocchi to the opendev.org project infrastructure. I agree with Adrian that it would facilitate getting openstack devs to work on it, but I agree with Julien that it\u0027s not a magic bullet either.\n\nThat said Adrian, if your end goal is to adopt it under the Telemetry team, I think you should open a thread about that on openstack-discuss. I\u0027m not convinced there is consensus over adding maintenance of a generic metrics storage database as part of \"openstack\".\n\nI personally think openstack should not be in the business of writing a database, so I\u0027m personally against adding such technical debt -- if Gnocchi cannot be relied on (be it maintained on GitHub or opendev.org), I\u0027d rather work on migrating to another DB than make Gnocchi an openstack service and collectively commit to maintaining it forever.","accounts_in_message":[],"_revision_number":1},{"id":"0ba87fb4c726fcefb112513d983ee30f5b8b0a3d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-25 23:41:03.000000000","message":"Patch Set 1:\n\n@Julien\n\nThe problem isn\u0027t that the project is on Github, the problem is that it appears to be a project that doesn\u0027t have much usage outside of OpenStack, and therefore few contributors from those sources. In OpenStack it is being used, and it likely would get more contributors if it was clear that this really was an \u0027OpenStack\u0027 project. But people like me are less inclined because essentially it is a requirement now for Ceilometer and OpenStack because Ceilometer no longer has an API, but it\u0027s not a service under our infrastructure or goverence. Gnocchi is our API, and it isn\u0027t under our goverence. That\u0027s just not ok for the long term health of Telemetry.\n\nYou didn\u0027t answer my question, do you know the metrics on non-OpenStack users of Gnocchi?\n\n@Thierry\n\nThe problem is that Ceilometer doesn\u0027t have an API anymore. When Gnocchi was built (while part of OpenStack) the Ceilometer API and a bunch of functionality was ripped out because \u0027Gnocchi covers that now\u0027.\n\nGnocchi is essentially a core service in OpenStack, but it\u0027s not in OpenStack because it was spun off. We are in a tough place in Telemetry because we NEED Gnocchi. Or we need to do a massive rewrite to return the API, or more likely make a new one. All while switching to a different backend that isn\u0027t Gnocchi, and hopefully isn\u0027t Mongo again...\n\nGnocchi is essential right now, but it\u0027s under external goverence. Which honestly never should have happened considering that it is the API for Telemetry. What should have happened is how Swift works standalone. The project maintains its ability to be run standalone, but remains part of OpenStack.\n\nI don\u0027t disagree that we don\u0027t particularly want to maintain a time series database, but Gnocchi does solve our issue with Telemetry fairly well in theory, but also happens to be our API right now.\n\nMaybe we can look at using InfluxDB, it looks like people kind of used it before:\nhttps://github.com/mcaimi/openstack-ceilometer-influx-publisher\n\nBut that stil means we need to write an API for things like Aodh, Cloudkitty and others to consume, and we\u0027d need to pick InfluxDB as the official backend (while still allowing other publish targets), and maybe massage the data in Ceilometer like Gnocchi does during collection to lower the granularity for older data when going to the \u0027built-in\u0027 backend of InfluxDB. But then we are essentially rebuilding Gnocchi around an external (and maintained) database.\n\nMoving Gnocchi back to OpenStack so we can actually control what is essentially the Telemetry API, and maintain it, is the least amount of effort right now, but yes then we are committing to mainain a time series database.\n\n@Julien\n\nI like your work, but you can see how you\u0027ve put us in an awkward situation. :(","accounts_in_message":[],"_revision_number":1},{"id":"5fd43c56a2ce681552392b0feffdc3757094bbcb","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-08-26 07:04:26.000000000","message":"Patch Set 1:\n\n\u003e You didn\u0027t answer my question, do you know the metrics on non-OpenStack users of Gnocchi?\n\nNo, sorry, I don\u0027t have such data, unfortunately.","accounts_in_message":[],"_revision_number":1},{"id":"0b978a0fff305942ed84b5ef05cf381d515fa89e","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-26 13:01:47.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Like I said above, this change is not discussing \"returning gnocchi\n \u003e to openstack\", it\u0027s just discussing returning Gnocchi to the\n \u003e opendev.org project infrastructure. I agree with Adrian that it\n \u003e would facilitate getting openstack devs to work on it, but I agree\n \u003e with Julien that it\u0027s not a magic bullet either.\n \u003e \n \u003e That said Adrian, if your end goal is to adopt it under the\n \u003e Telemetry team, I think you should open a thread about that on\n \u003e openstack-discuss. I\u0027m not convinced there is consensus over adding\n \u003e maintenance of a generic metrics storage database as part of\n \u003e \"openstack\".\n \u003e \n \u003e I personally think openstack should not be in the business of\n \u003e writing a database, so I\u0027m personally against adding such technical\n \u003e debt -- if Gnocchi cannot be relied on (be it maintained on GitHub\n \u003e or opendev.org), I\u0027d rather work on migrating to another DB than\n \u003e make Gnocchi an openstack service and collectively commit to\n \u003e maintaining it forever.\n\n+1 to everything Thierry said. If we\u0027re seeing Gnocchi as a blocker, or if it\u0027s becoming one, we should work to break away from it assuming these blocker issues can not be overcome. Otherwise we should treat Gnocchi as an external dependency and work with the project on their chosen development platform, in this case we\u0027re talking about github; this is similar to what other projects do with tools such as Ansible, Puppet, Chef, etc.","accounts_in_message":[],"_revision_number":1},{"id":"4cc127bcd7525ab62115d19c6ff44f6a149120a9","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-08-26 21:21:21.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e +1 to everything Thierry said. If we\u0027re seeing Gnocchi as a\n \u003e blocker, or if it\u0027s becoming one, we should work to break away from\n \u003e it assuming these blocker issues can not be overcome. Otherwise we\n \u003e should treat Gnocchi as an external dependency and work with the\n \u003e project on their chosen development platform, in this case we\u0027re\n \u003e talking about github; this is similar to what other projects do\n \u003e with tools such as Ansible, Puppet, Chef, etc.\n\nI totally disagree with things said above.\n\nFirst, Gnocchi is not a project like Ansible or Puppet nor a simple DB, like Adrian said, Gnocchi is kind of \u0027Ceilometer API\u0027, and it\u0027s consumed by other OpenStack projects, it\u0027s quite important for features like auto-scaling or auto-healing in OpenStack.\n\nGnocchi is a blocker right now because of no maintainers, but why not we solve the problem together but re-create another wheel? Why do you assume this blocker can\u0027t be overcome given what Adrian is trying to do here is actually moving Gnocchi development back on track? The people who chose the Github development platform have all left, which proves how a bad idea it was to split projects from OpenStack, why we still have to stick with the wrong way?\n\nMoving Gnocchi back into OpenDev makes it easier to integrate with other OpenStack services in CI and easier for OpenStack developers to get involved, and easier for the project governance.\n\nAdditionally, I know there are some people already deployed Gnncchi in production, I believe getting Gnocchi back on track is also good news for them. (I will try to find their names and add them for review.)","accounts_in_message":[],"_revision_number":1},{"id":"f1275b934516a659ea155f167ee36781d6728dfa","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-27 03:56:03.000000000","message":"Patch Set 1:\n\nAs Lingxian has reiterated, Gnocchi is the \u0027Ceilometer API\u0027. That\u0027s the core part of our whole problem right now.\n\nThat essentially makes Gnocchi an official part of OpenStack whether we like it or not because it is directly consumed by other services AS IF it is an official API for Telemetry, and it is capable of using Keystone for Auth.\n\nCan we change that? Sure, but that means \u0027still\u0027 maintaining Gnocchi while we write a new API back into Ceilometer, and choose a new backend for that API which can\u0027t just be a plugin export target for Ceilometer but has to be fully integrated (or a plugin backend system). All while migrating everything that consumes Gnocchi to the new API.\n\nThat\u0027s a HUGE amount of work, and still we have to somehow keep Gnocchi alive while we do that. Where will we find the devs for that?\n\nTelemetry is in a terrible situation because of Gnocchi leaving OpenStack, and because of the Ceilometer API being deprecated. It sucks to say, but that\u0027s the hard and sad truth of it.\n\nEven if we don\u0027t want to maintain a times-eries database (and trust me, I don\u0027t), bringing Gnocchi back into the fold is the easiest thing to do, despite the fact that we are then stuck maintaining a time-series DB.","accounts_in_message":[],"_revision_number":1},{"id":"76d7acc0bb4e471ee490691f63e16e30fb8f7073","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-08-27 06:16:17.000000000","message":"Patch Set 1:\n\nA benefit I\u0027d see by moving this back to OpenDev is that we\u0027d catch issues like the breaking changes with pecan 1.4, see https://github.com/gnocchixyz/python-gnocchiclient/pull/110 earlier.\n\nI\u0027m not sure how moving it back to OpenDev would help with issues like CI being broken (https://github.com/gnocchixyz/python-gnocchiclient/pull/110/checks ).\n\nGnocchi is integrated tightly into heat and aodh and these components provide features like autoscaling for OpenStack. Unlike other projects like libvirt or mysql/mariadb, Gnocchi is rarely(?) used outside of OpenStack scope.\n\nThe more user visible change would be: gnocchi is back under the OpenStack roof, and it may imply it is supported by the OpenDev community. The latter would be false, if there are no people stepping up.\n\n\nTaking a step back, since gnocchi is currently integral part of OpenStack autoscaling, but at the same time, the contributions to that part becoming less and less, we may have to either step up and contribute, find a replacement, or to accept that this feature is gone and users will see a regression.","accounts_in_message":[],"_revision_number":1},{"id":"eb287a93c304e75bcff799d123c5327eca392629","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2020-08-27 06:34:18.000000000","message":"Patch Set 1: Code-Review+1\n\nLeaving architectural changes aside (re. add API back to ceilometer), personally I do consider gnocchi being a part of OpenStack telemetry. \n\nThus I am supporting this step to move gnocchi back to OpenStack. The telemetry team will have to come up with a long-term plan: either get a replacement or to get this all into a more sustainable status.","accounts_in_message":[],"_revision_number":1},{"id":"349c595160c7e72389e345e2f8611c0e83acbf1b","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-08-27 07:54:23.000000000","message":"Patch Set 1: Code-Review-1\n\nLet me take a step back because I think some of you do not have a grasp on how some parts of the open-source dynamic works.\n\nNone of you ever contributed a single line of code to Gnocchi. So, with all due respect, what you think is completely irrelevant to any of the Gnocchi historical contributors. Understand that we don\u0027t know who you are and you are complete strangers talking on the Internet about our project. You have opinions? Fine. Everybody has those. They\u0027re cheap.\n\nYou can talk all you want in this review about what you think Gnocchi should be or whatever. Be my guest.\n\nIf you want leverage over an open-source project, you have to be part of it (\u003d contribute). Until you do that, you can talk all you want, you\u0027re going to be mostly ignored.\n\nThe current situation is totally fine in the eyes of Gnocchi\u0027s current development team and contributors. We\u0027re hosted on the largest open-source development platform. We\u0027re opened to contributions.\n\nIf you spent half the energy you spent writing in this code review contributing to Gnocchi, you\u0027d be way more efficient on the future of the project.\n\nGood luck!","accounts_in_message":[],"_revision_number":1},{"id":"0849670310e42764f013166ede25bcf0da55e334","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-27 08:09:58.000000000","message":"Patch Set 1:\n\nI\u0027ll try to stay neutral, but for the love of *something* please not another API or reinvention of the wheel.\n\nPersonally I don\u0027t like working on GitHub, but I have atleast tried starting to get CI working tho with extremely little time and very little knowledge into the project.\n\nWhile it would make all our lives easier moving it to OpenDev (it sure would for me, using OpenDev CI would be much better imo) I dont see why we can\u0027t step up and atleast get the upstream project\u0027s CI working, check reported bug and pending PR\u0027s and atleast start doing releases when bugs are fixed?\n\nWe are multiple people that could easily get it into a (atleast more) healthy state.","accounts_in_message":[],"_revision_number":1},{"id":"6cb2bd375af129ed8fae3830d1d6c7c29772a561","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-27 08:10:34.000000000","message":"Patch Set 1:\n\nShould also mention we run it in production, we know third party software (one of which we use) that support the Gnocchi API.","accounts_in_message":[],"_revision_number":1},{"id":"0abfb8028ae4ffb894319349e64868f68b7db44e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-27 13:26:45.000000000","message":"Patch Set 1: Code-Review-1\n\nFor me, gnocchi is a an external tool, like many other tools or libraries OpenStack consumes. It\u0027s fine for me to have it separately. I am not against bringing it in OpenStack, but I want it to be done for the right reasons. I am not sure it\u0027s the case here.\n\nIs there any technical issue preventing github contributions? AFAIK, it\u0027s accessible worldwide. If it\u0027s just about convenience, and multiple contributors don\u0027t want to contribute on github on the original project, I find it sad. So far, Julien was not against any contributions. We should build a stronger ecosystem, not fragment it even more.\n\nIf people miss Zuul for the project, maybe we can contribute to gnocchi and ask kindly if it\u0027s possible to add extra testing with Zuul on github. AFAIK, it\u0027s not impossible.\n\nI don\u0027t want to prevent people from contributing into the project by -1 this forever. But I don\u0027t want to fork (and build future tech debt) out of pure convenience... Let\u0027s work together folks :)","accounts_in_message":[],"_revision_number":1},{"id":"7296b86ba7ad5176405d4777a943b9fe25734898","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-27 14:42:39.000000000","message":"Patch Set 1:\n\nAdrian Turjak, Lingxian Kong, can you discuss with Julien about your concerns and discuss your eventual contributions, and report here, before we take this road of a hard fork? That would be lovely.","accounts_in_message":[],"_revision_number":1},{"id":"badc1e41cb04383555423e13cb2a4310e21e4a4c","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-08-27 15:00:05.000000000","message":"Patch Set 1:\n\n\u003e The current situation is totally fine in the eyes of Gnocchi\u0027s current development team and contributors. We\u0027re hosted on the largest open-source development platform. We\u0027re opened to contributions.\n\nSo I think we have a communication problem here, because this is very different from the message that people are receiving on GitHub, which is:\n\n* The current development team and contributors do not exist: \"nobody\u0027s usually around\" [1]\n* Pull requests will be ignored: \"[Gnocchi] is unmaintained\" [2]\n\n[1] https://github.com/gnocchixyz/gnocchi/pull/1059#issuecomment-607059270\n[2] https://github.com/gnocchixyz/gnocchi/issues/1049#event-2712977010","accounts_in_message":[],"_revision_number":1},{"id":"cd4f8e321c5c06bac92f42111dcfd0708cf86821","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-27 15:17:13.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e \u003e The current situation is totally fine in the eyes of Gnocchi\u0027s\n \u003e current development team and contributors. We\u0027re hosted on the\n \u003e largest open-source development platform. We\u0027re opened to\n \u003e contributions.\n \u003e \n \u003e So I think we have a communication problem here, because this is\n \u003e very different from the message that people are receiving on\n \u003e GitHub, which is:\n \u003e \n \u003e * The current development team and contributors do not exist:\n \u003e \"nobody\u0027s usually around\" [1]\n \u003e * Pull requests will be ignored: \"[Gnocchi] is unmaintained\" [2]\n \u003e \n \u003e [1] https://github.com/gnocchixyz/gnocchi/pull/1059#issuecomment-607059270\n \u003e [2] https://github.com/gnocchixyz/gnocchi/issues/1049#event-2712977010\n\nWell it\u0027s a different problem if the project is not maintained! It\u0027s not like it\u0027s two groups conflicting. It\u0027s just one single group wanting to maintain the way they want, and chosing their hosting platform. I would not call this a \"return\", but instead a \"fork\" of an unmaintained the project. That\u0027s a good news for the open source world then.\n\nReverting my vote.","accounts_in_message":[],"_revision_number":1},{"id":"4f2594fdbb1cb8628284449efca1fce700968849","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-27 15:17:20.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"9e08cb23fd837c0fe5f55458a158bb7c0cdfed0c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-27 15:21:13.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"27bd5a4d869ff56484b2db678a2796c1566c6888","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-27 15:39:13.000000000","message":"Patch Set 1:\n\nFor folks interested in maintained the project under OpenDev, I encourage you.\n\nPlease respin this with a new name for the project and under a seperate namespace (i.e.: foobar/foobar) as it doesn\u0027t make great sense to be inside OpenStack (for now at least).\n\nI would reword the commit message to explain that this is a fork of Gnocchi.  I\u0027m happy to push forward with this after","accounts_in_message":[],"_revision_number":1},{"id":"7cb9758e2c03abff921836316253188622fdbf8d","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-08-27 16:09:37.000000000","message":"Patch Set 1:\n\nWith the story on GitHub being that the project is unmaintained, a fork is reasonable. I agree with JP and Mohammed that this needs a new namespace and name. (It won\u0027t be under OpenStack governance, and to prevent confusion/trademark issues.) +1 after that.","accounts_in_message":[],"_revision_number":1},{"id":"92ef1099e74480e8047be7d061f4c1fd406aaa68","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2020-08-27 16:23:49.000000000","message":"Patch Set 1:\n\n(2 comments)\n\n\u003e \u003e The current situation is totally fine in the eyes of Gnocchi\u0027s\n \u003e current development team and contributors. We\u0027re hosted on the\n \u003e largest open-source development platform. We\u0027re opened to\n \u003e contributions.\n \u003e \n \u003e So I think we have a communication problem here, because this is\n \u003e very different from the message that people are receiving on\n \u003e GitHub, which is:\n \u003e \n \u003e * The current development team and contributors do not exist:\n \u003e \"nobody\u0027s usually around\" [1]\n \u003e * Pull requests will be ignored: \"[Gnocchi] is unmaintained\" [2]\n \u003e \n \u003e [1] https://github.com/gnocchixyz/gnocchi/pull/1059#issuecomment-607059270\n \u003e [2] https://github.com/gnocchixyz/gnocchi/issues/1049#event-2712977010\n\nI whole heartedly agree with spirt of what Julien is saying: folks should work with the Gnocchi community upstream in their chosen development platform, it\u0027s just like all other OpenSource software on github, and their community is emphatically stating \"pull requests are welcome.\" The Gnocchi project made a conscious decision to move away from the OSF and we should respect that. All that said, I would support forking if that\u0027s the actual desire; which Jean-Philippe captured in comments.\n\n \u003e If we bring it back into our fold it makes it easier to maintain and potentially rally people\n \u003e to work on it, because then the OpenStack community collectively holds governance over it and\n \u003e if the project shows a lack of development effort it makes it easier for someone else to take over.\n \nInitially I was under the impression that the stated desire was to bring development back in-house for an easier developer experience, but if that\u0027s not the case, this commit message should be updated to state the actual desire, purpose, and plan.","accounts_in_message":[],"_revision_number":1},{"id":"706fdeceb6c33f923fdd658a85197464388939e1","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2020-08-27 19:05:56.000000000","message":"Patch Set 1: Code-Review-1\n\nI agree that we need to make a distinction with a fork and a new name based on the conversations that were had in IRC today.","accounts_in_message":[],"_revision_number":1},{"id":"b1334cd9fc8302a6bba2fcef526c2fb903e45df0","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 00:06:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"004a2b6f7a4b252f713c57e93fe3b95f847af567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 00:13:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/59496de04ea143f39f21a853b60100a2 : SUCCESS in 6m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4f634d5162654ef48cef4edf349cf769 : FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9ba4fd696e209a851ff27b8bcb1148c463a1626d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 00:19:21.000000000","message":"Patch Set 2:\n\nI\u0027d like to stress that if we are not bringing Farfalle (you are welcome to suggest a better name :P) back under the \u0027openstack\u0027 namespace eventually, we should commit to getting rid of it as a whole and moving to a different backend and building a v2 API into Ceilometer.\n\nI know Jean and others are against it, but unless we commit to maintaining our fork under the official namespace, Gnocchi SHOULD NOT be the official telemetry API. But given that it is right now... we very soon need to make that choice. Do we maintain it long term, or commit to replacing it.\n\nWe don\u0027t need to do that right now, but I will likely end up writing an email to the list about our options once this ends up back on opendev.\n\nI\u0027ve opted for \u0027openstack-telemetry\u0027 as the namespace so it sounds semi-official and isn\u0027t just \u0027farfallezyx/farfalle\u0027. It\u0027s ultimately just a placeholder until we move it officially to openstack, or write a replacement API and move to a different backend.","accounts_in_message":[],"_revision_number":2},{"id":"124069ad7ff92c3df761e9280fece0ac38c6adfb","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-08-28 01:51:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c77c79886d61e54df4f661693040a42927e2787e","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 02:42:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"786cfd8bb3d36aebe4f551b212a0642cd79634c0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-28 08:44:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f773233901c950c3ea197f9c3ed762b11fa1e077","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-08-28 08:48:29.000000000","message":"Patch Set 2:\n\n(and I also still think we need to raise this topic on the mailing-list,  so that this review is visible to more than infra, the TC and JD\u0027s Twitter followers).","accounts_in_message":[],"_revision_number":2},{"id":"1c97fee86df3d339cb3e202ae50ed836bce68e75","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-28 11:32:29.000000000","message":"Patch Set 2:\n\nI would be in favour of placing it under farfelle/farfelle too. We can rename later.","accounts_in_message":[],"_revision_number":2},{"id":"85a8df007b347daa21450a65feae1c32d5b9a176","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-28 11:32:58.000000000","message":"Patch Set 2:\n\n\u003e (and I also still think we need to raise this topic on the\n \u003e mailing-list,  so that this review is visible to more than infra,\n \u003e the TC and JD\u0027s Twitter followers).\n\nAgreed","accounts_in_message":[],"_revision_number":2},{"id":"cbcac563176a31305204245b7b0de49e4c46145a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-08-28 11:35:26.000000000","message":"Patch Set 2:\n\n\u003e \u003e (and I also still think we need to raise this topic on the\n \u003e \u003e mailing-list,  so that this review is visible to more than infra,\n \u003e \u003e the TC and JD\u0027s Twitter followers).\n \u003e \n \u003e Agreed\n\nAgreed on the not making an official project, but a side project unrelated to openstack (as this is generic). We thank the contributors for willing to host the new project in the opendev ecosystem, close to openstack friends :)","accounts_in_message":[],"_revision_number":2},{"id":"b615d182bd0bd951bba78344a4f5c47c51a3b0e4","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 12:06:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d75390d2c13d5e435976ec3febf4a179c9b49631","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 12:16:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/90b983458cd945d380209663f8db7943 : SUCCESS in 7m 09s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d0db8d56f2d4208b44d2762afdce79a : FAILURE in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"44f3e0bd027c945c361b4d8f72f04cd91b2c2413","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-28 12:26:52.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nPlease follow:\nhttps://docs.opendev.org/opendev/infra-manual/latest/creators.html\n\nThis needs at least a change for zuul/main.yaml as well.","accounts_in_message":[],"_revision_number":3},{"id":"d55a57a0542fefdf9edf12afe6d40669c9be6235","author":{"_account_id":2813,"name":"Mehdi Abaakouk","email":"sileht@sileht.net","username":"sileht"},"date":"2020-08-28 13:05:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"deb940dc23b31802f8177a8897f8ee5e31ec2766","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 13:22:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5904349159d05546437427c181c935ea05d95bc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:34:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/236afc72d405465e80a9cef327f7fee5 : SUCCESS in 5m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3ffcafc3dece4f5b83bf17a30e378dc8 : FAILURE in 3m 40s","accounts_in_message":[],"_revision_number":4},{"id":"583c8c7acb2a1fc533faab588ae37825dfd428dc","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 13:48:25.000000000","message":"Patch Set 4:\n\n@Medhi \n\nThanks for the comment.\n\nI think we would prefer not to outright fork it, but our issue is that it seems OpenStack are the only users of Gnocchi, and it is pretty much our official API for telemetry. The reason I personally don\u0027t want to contribute to Gnocchi as it is on github is that the project is under external governance. I personally dislike gerrit, but as an OpenStack dev who knows that Gnocchi is relied on by us, I feel that we need to be able to drive the direction of the project, and have some governing body that can step in an appoint new maintainers if the project dies again.\n\nI have nothing against you, Julien, or Gordan, but trying to revive a project that only we use while it is not fully under our control from a governance perspective is essentially broken as a concept. A fork in that respect is cleaner, especially since we don\u0027t even know if we want to keep Gnocchi around yet.\n\nIdeally we would have worked with you to move the project back to OpenStack and maintain it as a standalone service like Swift.\n\nBased on my understanding for why Gnocchi left OpenStack, it was to try and get support from external contributors outside of just OpenStack who didn\u0027t want to use Gerrit. I get that, but it clearly failed, and in the process you lost a lot of OpenStack contributors.\n\nWe just want a solution, and given it seems neither yourself, Julien, or Gordan have time for Gnocchi anymore due to other projects and work, someone does need to take over the project, but for us (OpenStack as a whole) it doesn\u0027t make sense to do that via github and your project group unless that group and contributors was already active. Either we are contributing to an active project (which it isn\u0027t) or it simply makes more sense to move the project back somehow and just deal with it entirely under our own control.","accounts_in_message":[],"_revision_number":4},{"id":"bb49940d4185456aff446c204c940d77693a7dd0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 13:57:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)\n\n-1 for the formal comments; name comment is not blocking but I would love to see it reflect the original name better :-)","accounts_in_message":[],"_revision_number":4},{"id":"9d2abfe2ec04d46c1e2f002e07dd2761f47ea610","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 13:59:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c8cb2824cd920165144fb60a00e597f13e0b1986","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-08-28 14:01:15.000000000","message":"Patch Set 4:\n\nAdrian, you have things upside down.\n\nWhat Gnocchi needs is developers, not a governing body.\n\nIf you think changing the \"governance\" is going to bring developers, good luck.\n\nAlso, we didn\u0027t lose any contributors moving to GitHub, you got things wrong on that. We already had no contributor on OpenStack, haha.","accounts_in_message":[],"_revision_number":4},{"id":"13f6f7138108d7d1998583db6347b61a8d3183f9","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 14:02:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"922c8a76244f176171c4b45949ce4611da3da723","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 14:12:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/73012255634842bbace5f83fcb8adb7c : SUCCESS in 4m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b02d39084263442caeb7680c84c273a2 : FAILURE in 3m 36s","accounts_in_message":[],"_revision_number":5},{"id":"472e30d322d0da7da807e612a74f11278034c828","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 14:15:54.000000000","message":"Patch Set 5:\n\n\u003e What Gnocchi needs is developers, not a governing body.\n \u003e \n \u003e If you think changing the \"governance\" is going to bring\n \u003e developers, good luck.\n\nNo, I think changing the project leadership will allow us to actually merge fixes to the CI while dealing with known bugs, and then decide what to do with the project going forward.\n\nThere are people willing to contribute as far as I\u0027m aware, but based on what I see on Github, and your attitude here, I don\u0027t see anyone willing to lead the project for its benefit. We want a solution to keep the project alive, and mostly what I seem to have seen from you is sarcasm and obstinacy.\n\n \u003e Also, we didn\u0027t lose any contributors moving to GitHub, you got\n \u003e things wrong on that. We already had no contributor on OpenStack,\n \u003e haha.\n\nI assumed as much, but I was being optimistic and generous. Clearly moving to Github didn\u0027t fix that though.","accounts_in_message":[],"_revision_number":5},{"id":"8563287cab7533e6b39a7282911006d9a55c0b36","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 14:32:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2ed6691f322d29b23f57a9a588bd21529b0163d8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-28 14:33:23.000000000","message":"Patch Set 5:\n\nalso, to past tests you need a bit more dance, see my etcd3gw move: https://review.opendev.org/747185","accounts_in_message":[],"_revision_number":5},{"id":"2289c241c5f9306446a5786dcafe4f2b2c8ac352","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-08-28 14:44:08.000000000","message":"Patch Set 5: Code-Review-1\n\nAs one of the people involved in decomposing the Ceilometer monolith into Aodh and the various other parts, before but eventually leading to Gnocchi, there\u0027s some history here that may not be obvious.\n\nThose of us who were working on Ceilometer at the time were doing so because we were paid to do so, not because it was reasonable for it to exist as an \"OpenStack\" project. And the numbers of contributors kept shrinking, despite a fairly constant rate of vendors and using-clouds asking for features without actually making contributions.\n\nEven then it was obvious that telemetry in its various forms (metrics, monitoring, alarms, rating, billing) could be sufficiently generic to be supported by third party software that had active communities. It would have made more sense to improve the existing OpenStack services so that they produced or exposed the necessary metrics but the politics and preferences of the time (2014-15) made that difficult.\n\nPart of the point of doing the decomposition was a) to make it easier to comprehend and contribute to a messy system, b) (more importantly) to make it more obvious that the telemetry projects were inconsequential and should be thought of roles that should be filled with tools, in the same that MySQL or PostgreSQL can fill the RDBMS role.\n\nHistory bears out the above. The choices for third party open source telemetry are pretty good and if OpenStack can\u0027t work with them, that\u0027s OpenStack\u0027s problem, one that would be much more useful to solve than debating the future of Gnocchi.\n\nAs said elsewhere: If people need or want to use Gnocchi, contribute to it. If you refuse to use Github, get over yourself. When I joined the OpenStack community I _really_ did not want to use gerrit but came around because it was more important to make contributions.\n\nI mean no disrespect to Adrian with my response here, because he\u0027s done plenty of good things in the community, but I think it is terribly shortsighted to believe that a fork Gnocchi to a different location will do anything to change the contribution profile. It appears moribund because it is moribund and it will remain so wherever it lives.\n\nThe only way to change that situation, if Gnocchi really is required, is to treat it and attend to it like it is required. If that hasn\u0027t happened in the last four years, why is now any different?\n\nIt probably isn\u0027t. Try a different strategy, use tools.","accounts_in_message":[],"_revision_number":5},{"id":"3be11c9096c2516acad90582a72ff1a83a5105de","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-28 14:57:28.000000000","message":"Patch Set 5: Code-Review-1\n\nLet\u0027s try to collectively give some love to the upstream Gnocchi project.\n\nTo build upon Chris comment I\u0027ve heard a lot in the past about Ceilometer being built around technical issues (read: polling projects APIs for usage) instead of there having been a discussion (due to said politics back then) on how to collectively solve it in OpenStack and standardize.\n\nJust replacing Gnocchi (not just the API, but storage backend as well) is also a big discussion, seeing as how there were multiple backends for Ceilometer back then but it never really worked properly.","accounts_in_message":[],"_revision_number":5},{"id":"fc9bdc19fd1cf9ab45faefb0f1aeae03f681b52f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-28 15:06:27.000000000","message":"Patch Set 5:\n\nEven if I\u0027d wanted to contribute to gnocchi to keep it alive, I have no chance in doing that! As PR are just hanging there without having any opportunity to merge.\n\nAnd I know at least several public cloud providers that rely on openstack\u0027s telemetry as it\u0027s simple to install, reliable and goes out of the box.\n\nSo I\u0027d say that moving gnocchi under opendev is more about giving an opportunity to contribute and having a contact person who can review PR and approve some patch, which does not happen on the Github from what I see.","accounts_in_message":[],"_revision_number":5},{"id":"e62735d045af7cdb6137f82330d31ea5e5dbd38f","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 21:15:18.000000000","message":"Patch Set 5:\n\n@Chris\n\n \u003e History bears out the above. The choices for third party open\n \u003e source telemetry are pretty good and if OpenStack can\u0027t work with\n \u003e them, that\u0027s OpenStack\u0027s problem, one that would be much more\n \u003e useful to solve than debating the future of Gnocchi.\n\nWe have no choice but to debate the future of Gnocchi until we have a suitable replacement because people rely on it and its API. It\u0027s not just as simple as saying: \"Prometheus is the answer!\" and calling it a day, because that also doesn\u0027t solve all the use cases, and it isn\u0027t a simple or quick migration for anyone.\n\n \u003e As said elsewhere: If people need or want to use Gnocchi,\n \u003e contribute to it. If you refuse to use Github, get over yourself.\n \u003e When I joined the OpenStack community I _really_ did not want to\n \u003e use gerrit but came around because it was more important to make\n \u003e contributions.\n\nNo one is refusing to commit to Github, but the group on Github has 3 people, none of which are active on the project anymore. Which means we would be contributing in hopes they review our code and merge it, but for people who have tried, that hasn\u0027t happened. Unless Julien and co and confirm their intent to review and lead the project, then it wouldn\u0027t make sense at all to continue on Github unless they those interested in maintain Gnocchi control of the Github group. But given that only OpenStack seems to use Gnocchi and most of the devs interested in supporting it are here, moving it back to our infra IF we are the only ones maintaining is is smarter than trying to deal with core devs who aren\u0027t responsive.\n\nI doubt the people who control the Gnocchi group actually run Gnocchi in prod (I could be wrong), but a lot of OpenStack people do, and we therefore have a vested interested in keeping it alive, even if just long enough to replace it with a maintained third party TSDB which we wrap with a stable API for internal use cases (because such an abstraction is sadly needed)\n\nIf it was active on Github, we wouldn\u0027t be trying to fork it, and Julien was sarcastic off the bat when I opened this review to get the discussion started. Which I did because Github was silent, and no one knew what the hell to do anymore about this mess.\n\n \u003e I mean no disrespect to Adrian with my response here, because he\u0027s\n \u003e done plenty of good things in the community, but I think it is\n \u003e terribly shortsighted to believe that a fork Gnocchi to a different\n \u003e location will do anything to change the contribution profile. It\n \u003e appears moribund because it is moribund and it will remain so\n \u003e wherever it lives.\n \u003e \n \u003e The only way to change that situation, if Gnocchi really is\n \u003e required, is to treat it and attend to it like it is required. If\n \u003e that hasn\u0027t happened in the last four years, why is now any\n \u003e different?\n\nI don\u0027t believe simply moving it back to OpenStack with a fork will magically solve our problems, and I\u0027ve already said this a few times.\n\nWe are doing this because there are devs willing to contribute who want to actually have their code merged into the project, and that isn\u0027t happened on Github. This whole thing is us trying to fix this mess and actually support telemetry so it doesn\u0027t fall over. It won\u0027t be easy, but it needs to happen because too many people rely on it. We may well end up ripping large chunks of it all out, but right now getting it all back into a working state so we can start planning for the future roadmap is more important.\n\nThe first step is ACTUALLY being able to merge fixes and get a stable and running CI.","accounts_in_message":[],"_revision_number":5},{"id":"9836acf551a7e3950b98e270d4ca40854dfb5a0d","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 21:29:49.000000000","message":"Patch Set 5:\n\n\u003e Let\u0027s try to collectively give some love to the upstream Gnocchi\n \u003e project.\n \u003e \n \u003e To build upon Chris comment I\u0027ve heard a lot in the past about\n \u003e Ceilometer being built around technical issues (read: polling\n \u003e projects APIs for usage) instead of there having been a discussion\n \u003e (due to said politics back then) on how to collectively solve it in\n \u003e OpenStack and standardize.\n \u003e \n \u003e Just replacing Gnocchi (not just the API, but storage backend as\n \u003e well) is also a big discussion, seeing as how there were multiple\n \u003e backends for Ceilometer back then but it never really worked\n \u003e properly.\n\nAgain, we\u0027ve love to, provided the existing github group will continue to support the project and contributors. But given that they have other priorities (which happens and is fair enough) what I think we need is a change of leadership. So either they give us access to the Github group or (considering we already all mostly work here) we move it to opendev if the current devs (and reason it is on github) don\u0027t intend to continue with the project.\n\nNo one is saying we refuse to work on Github, what we are saying is that there doesn\u0027t seem to be a point.","accounts_in_message":[],"_revision_number":5},{"id":"e61f023360ff2d2b4426e8122e658ee65c4da3a9","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-28 21:51:11.000000000","message":"Patch Set 5:\n\nI picked Farfalle because it\u0027s a fun shape (squeezed like the data gnocchi processes), fun to say, and also Italian.\n\nGnocchi itself already mixes different Italian dishes, so that seemed more fitting:\nhttps://gnocchi.xyz/operating.html?highlight\u003dcarbonara#how-to-plan-for-gnocchis-storage\n\nKopytka, kartoffelknödel, etc, are all valid options.\n\nI genuinely don\u0027t care what we call it, as long as it works. :)","accounts_in_message":[],"_revision_number":5},{"id":"6b0b400c7c8c1d76b3ef78335e1e3f1fe6885e42","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-08-29 07:40:22.000000000","message":"Patch Set 5:\n\n\u003e I picked Farfalle because it\u0027s a fun shape (squeezed like the data\n \u003e gnocchi processes), fun to say, and also Italian.\n \u003e \n \u003e Gnocchi itself already mixes different Italian dishes, so that\n \u003e seemed more fitting:\n \u003e https://gnocchi.xyz/operating.html?highlight\u003dcarbonara#how-to-plan-for-gnocchis-storage\n\nYour explanation appeals to me! Thanks!\n\n \u003e I genuinely don\u0027t care what we call it, as long as it works. :)\n\nAs long as we don\u0027t call it Spaghetti! :-)","accounts_in_message":[],"_revision_number":5},{"id":"ec4280abd93fe5c4ff9fefef32a40ef3cdc0e5ce","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-29 11:56:51.000000000","message":"Patch Set 5:\n\nI don\u0027t agree though, the people on Gnocchi still watch the changes, but they don\u0027t want to spend the time to for example get the CI working.\n\nI don\u0027t see anybody stepping up to fix the CI in addition to said \"changes that is not reviewed\".\n\nI\u0027m certain that if we step up, get CI working, make sure pending PR\u0027s is fine and can be merged, they will get merged (given they follow policy). If we get some people working on it more actively I\u0027m sure those persons will get merge rights.","accounts_in_message":[],"_revision_number":5},{"id":"85b67645102d783e77fa4633e8380e395bd02838","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-29 11:58:23.000000000","message":"Patch Set 5:\n\nWe should also not understate the amount of surrounding work forking it would introduce, packaging, deployment projects, everything would be affected. But down to the core I understand the issues and frustation but maybe we can workaround needing to fork in the first place","accounts_in_message":[],"_revision_number":5},{"id":"d915c78616f03940946e98b88e5957c8637ef258","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-08-29 12:04:10.000000000","message":"Patch Set 5:\n\n\u003e I\u0027m certain that if we step up, get CI working, make sure pending\n \u003e PR\u0027s is fine and can be merged, they will get merged (given they\n \u003e follow policy). If we get some people working on it more actively\n \u003e I\u0027m sure those persons will get merge rights.\n\nYes, and if that doesn\u0027t happen, the thing to do then is do an actual fork. Same name, on opendev (but not OpenStack), new CI, that acknowledges that it is a fork and the reasons why. If you succeed you will no longer be the fork.\n\nAll this noise about naming, governance and association with OpenStack is secondary to the real needs. Just do open source. Don\u0027t do this bureaucratic encumbered thing that OpenStack has normalized.","accounts_in_message":[],"_revision_number":5},{"id":"e9d05b8da374614de23e63260286fd73aa40cd18","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-08-29 14:45:50.000000000","message":"Patch Set 5:\n\nOK then, a couple of straightforward questions:\n\nJulien, do you have any intention of continuing development with Gnocchi? That includes active contribution, design/feature planning, bug triage, helping review pull requests, and onboarding new devs to the project (both by working with them directly, and eventually adding them to the github group). Also, are you currently running Gnocchi in a production environment, or working with/for a company that does?\n\nMehdi, same questions. :)\n\n\n@Chris\n\nNo one wants a fork, but I feel like all I\u0027ve read from Julien is sarcasm and scorn towards anything (and anyone) OpenStack, without any intention of responding reasonably. Mehdi has been reasonable so I have more respect from him in this situation.\n\nI started this review to kick off a discussion based on what information I was given about the state of Gnocchi and from what I could see about the project contribution history on Github (and the dev responses there). It somehow very quickly devolved into whatever the hell this is.\n\nOpenStack needs Gnocchi to continue to function as an application for the foreseeable future, and there are people willing to contribute to it, but doing so on Github is pointless if the people who have the power to merge aren\u0027t themselves also active on the project. If not, then we fork it because we have to.\n\nThe rename is solely because we don\u0027t have control over the pypi targets, and similar, so a renamed fork is a clearer distinction and is common in forks (Icinga vs Nagios being an easy example, but not a good comparison to our situation at all). This review started with the same name because it was meant to be a starting point for: \"Julien, are you still maintaining this and doing reviews, or will you let us take over and bring it back to OpenStack?\" But it is now this whole bloody mess.\n\nSo let\u0027s get a definitive answer to the above, and then we can actually figure out what to do. Because given how Julien has responded here, and on Github (and Twitter), it feels like he doesn\u0027t care about Gnocchi anymore but also doesn\u0027t want to admit it.","accounts_in_message":[],"_revision_number":5},{"id":"b95f459e69891e2cc201e54aa27e7e279f4a2fe6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-31 07:21:48.000000000","message":"Patch Set 5:\n\n\u003e I don\u0027t agree though, the people on Gnocchi still watch the\n \u003e changes, but they don\u0027t want to spend the time to for example get\n \u003e the CI working.\n\nI\u0027m not sure tbh how valid it though. As an easy example that I was shocked with was gnocchi.xyz expiration, which put down docs for several month, and there was literally no reply from JD for month, even though ppl suggested to pay for domain.\n\nYou can easily check that with this issue https://github.com/gnocchixyz/gnocchi/issues/1048 \n\nAnd I still do run gnocchi docs mirror just in case:(\n\n \u003e I don\u0027t see anybody stepping up to fix the CI in addition to said\n \u003e \"changes that is not reviewed\".\n \u003e \n \u003e I\u0027m certain that if we step up, get CI working, make sure pending\n \u003e PR\u0027s is fine and can be merged, they will get merged (given they\n \u003e follow policy). If we get some people working on it more actively\n \u003e I\u0027m sure those persons will get merge rights.\n\nSo even if we step in and fix now, and PRs will be merged for some time (until hype related with this patch will lay down), I don\u0027t really believe in long lasting effect, as this case showed attitute to the project...\n\nSo at the moment we even don\u0027t have at least guaranteed docs and CI resources, so really can face up in expired domain or smth as trivial things at least.","accounts_in_message":[],"_revision_number":5},{"id":"0e3f7f1578eb98b3f33a335e38a02cd8a4a7aaf1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-09-01 01:49:12.000000000","message":"Patch Set 5:\n\n\u003e Gnocchi is clearly not actively maintain, but for sake of clarity: the pull requests have not been reviewed because the CIs have never passed.\n\nA lot of this seems to be about CI, which isn\u0027t working and an understandable blocker to upstream developers wanting to undertake code review.  On the other hand, nobody is actively invested in maintaining the upstream CI (based on reading the grab bag of required changes in https://github.com/gnocchixyz/gnocchi/pull/1065).\n\nAlthough we do not want OpenDev to be a generic TravisCI replacement, there are areas where it makes a lot of sense to have OpenDev Zuul CI running against external github projects.  Bringing together an inactive upstream CI effort with interested parties wanting to build out OpenDev CI would, to me, clearly be one. \n\nI have recently had a very productive interaction with pyca project wrt to cryptography where Zuul is providing CI resources for their project (this was not due to issues upstream, but because we have ARM64 resources that are particularly useful, and we get the benefits of having a stable arm64 environment).  Zuul reports in the same way as other CI systems using the checks api, see https://github.com/pyca/cryptography/pull/5428/checks for an example.\n\nUsing that as a template, we would have a separate gnocchi tenant for running the CI jobs in.  We can organise things in many different ways; we can create a project-config for pipeline configuration in the gnocchi OpenDev tenant, or upstream could create one and give people developing on Zuul access.  We can keep job config/playbooks/roles in the github gnocchi repository or separate repos.  At a minimum, all Gnocchi needs to do is enable the OpenDev github bot.\n\nThis can only work if a) upstream will not insist that the only CI is travisCI and b) that the people proposing the fork are committing to actually building out useful CI in Zuul.\n\nTo me, it seems like this might bridge the gap between where the active developers are working (OpenDev) and what upstream needs as a workflow to manage the project (CI showing changes aren\u0027t broken).\n\nIf people can come to some sort of agreement, which I have to honestly say seems unlikely given the way some of the discussion is going, I can commit helping get this to the point of running and reporting jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4287938c332bbb7b216688b386486b84c9a15cc1","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-09-01 10:57:09.000000000","message":"Patch Set 5:\n\nI don\u0027t run Gnocchi and don\u0027t work on it actively anymore since April 2019. I don\u0027t plan on actively writing code for it in the future if that answers your question.\n\nI\u0027ll be happy to help to the extent I can, reviewing PR for bug fixes for example. That\u0027s something we all tried to do for the last year — I know Gordon merged some bug fixes.","accounts_in_message":[],"_revision_number":5},{"id":"2bafa7b2b78cc7035ec3b479781c822954bef6aa","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-09-01 10:58:39.000000000","message":"Patch Set 5:\n\nJust forgot to reply to a point: I definitely care less about Gnocchi that I did when it was my job. I still feel like it\u0027s not a good idea to give the project to anyone randomly showing up on the Internet and asking for the keys. :)","accounts_in_message":[],"_revision_number":5},{"id":"f551828154f5db50aa6a4598b8013eef33298c50","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-09-01 11:02:19.000000000","message":"Patch Set 5:\n\nDmitriy, I feel like you misunderstood the issue with the domain.\n\nI bought the domain from my pocket a while back and asked my employer to take care of it so I wouldn\u0027t be a SPOF. I never managed to do so as nobody cared back then. Whatever.\n\nThe domain then expired, and when people asked to buy it back, it was impossible to do so for whatever reason.","accounts_in_message":[],"_revision_number":5},{"id":"03419436a60b9dd72087f0a1797bef7988e79100","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-09-03 03:44:06.000000000","message":"Patch Set 5:\n\n@Julien\n\nTy! That\u0027s what I assumed, but it\u0027s good to know. :)\n\nI know asking to be handed the keys when I personally have never contributed to Gnocchi is a bit of a stretch, but I genuinely am trying to keep the project alive (at least for the foreseeable future) and just want to facilitate a solution for those of us using it, or wanting to use it in OpenStack.\n\nOpenStack seem to be the only visible/vocal consumers of Gnocchi, so we (and people running OS, and being paid to maintain OS services upstream) have a collective reason to keep it alive. I\u0027m certain among those people we can find enough interest to help with bugs and new features etc. Once I have some current projects off my plate, I will be dedicating part of my time to helping with Gnocchi, so part of all this was me prepping a place to be able to contribute or potentially drive the project.\n\nMy reason for wanting to move it back to OpenStack infra isn\u0027t because I think that\u0027s the magic bullet solution to save the project, but because I felt that long term if you, Mehdi, and Gordan don\u0027t have good reasons to maintain the project, then it makes sense to hand it over to people who might. That way there is no burden on you to care/worry about it unless you want to, but the ultimate responsibility for the project ends up in OpenStack\u0027s hands. We have some silly but powerful infra, some complex but useful tooling, and processes to follow that while being a pain sometimes, do help govern the massive ecosystem that is OpenStack.\n\nThat\u0027s not to say that the TC or the foundation aren\u0027t sometimes unbelievably annoying bureaucracies to deal with, but at least if I (or someone else) fails as PTL for Gnocchi, they can step in and hand the project over to other interested parties. Which has happened a few times and sometimes it has allowed projects to be saved.\n\nSo while we can definitely work with you, Mehdi, and Gordan to rekindle the project on Github, that doesn\u0027t really make sense unless you do want to still take part in its future, or drive it. It doesn\u0027t make sense for us to contribute on github if we might never get access to or control of the gnocchixyz group from people who in our perspective don\u0027t have strong interest in the project anymore, and less so still if most of us are already all on opendev.org and gnocchi is mostly used by OpenStack. That\u0027s why if you don\u0027t see yourselves working on the project anymore seriously I think closing it off on github, and moving it back to OpenDev is the best case for its on-going future as a project.\n\nI can\u0027t say we will want to keep it around forever given that there is unlikely to be huge interest long term in maintaining a TSDB when we can potentially switch to using InfluxDB (or something else) as a backend. I don\u0027t expect that choice to happen soon, and if it does, we\u0027d still end up maintaining Gnocchi for a while to allow a transition. In either case, Gnocchi lives on for another few years, maybe longer. We may even find that the alternatives (Influx, etc) don\u0027t handle our cases, or the amount of data we ingest, and end up needing Gnocchi forever.\n\nSo please do think about it, because I really just want Gnocchi to live on as a project and I think this (minus the rename) is the best path forward for it.\n\nIf we end up deciding that for the future of OS-Telemetry we need something else, we\u0027ll be vocal about it early, and allow any interested parties to take over stewardship of Gnocchi outside of OpenStack, but until then we will keep it alive and maintained.","accounts_in_message":[],"_revision_number":5},{"id":"8999512729c30cb321a855d7da91e844b3d6fcd1","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2020-09-03 08:53:07.000000000","message":"Patch Set 5:\n\nThanks for the reply Adrian. I don\u0027t have much to add, I feel I\u0027ll keep repeating myself. :)\n\nWhen you write:\n\u003e then it makes sense to hand it over to people who might\n\nThis is exactly my point. You _might_ do a better job than us, or you _might not_, and that might just be a lot of fuss for nothing. We have no clue to know for now. And until now it has a lot of fuss for nothing. \n\nAs of today, the only person I\u0027d trust would be Tobias because they\u0027re actually trying to fix the CI and make progress while we debate about an hypothetical future. :)\n\nI cannot state that enough: if you want to move a project, come to it, gain the trust of the maintainers, get (a copy of) the keys, and steer it. When old grumpy maintainers get annoyed with young and fresh new maintainers doing much of the work, they just give more and more way. :)\n\n(I don\u0027t think you were around when the Ceilometer team started, but you\u0027d be surprised by the diversity and the number of core reviewers we onboarded compared to the size of other OpenStack projects.)\n\nIf you don\u0027t want to play that game for whatever reason or disagree with the maintainers, as many stated, go ahead and fork the project. I think it\u0027d be hard to justify a fork knowing that you didn\u0027t even try to send a single line of code to the project in your case — but it\u0027d not be my problem to justify such a decision.\n\nThis is how open-source works — outside of the OpenStack bubble. :)","accounts_in_message":[],"_revision_number":5},{"id":"3df983f0fa5f1a7a3b1c872913a0e29500dfb58a","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-09-04 03:37:44.000000000","message":"Patch Set 5: Workflow-1\n\n\u003e This is exactly my point. You _might_ do a better job than us, or\n \u003e you _might not_, and that might just be a lot of fuss for nothing.\n \u003e We have no clue to know for now. And until now it has a lot of fuss\n \u003e for nothing.\n \u003e \n \u003e As of today, the only person I\u0027d trust would be Tobias because\n \u003e they\u0027re actually trying to fix the CI and make progress while we\n \u003e debate about an hypothetical future. :)\n\nBut feels like it has taken a lot of this debate for his patches to get feedback/traction. :(\n\nWe can always make him PTL OpenStack side. :P\n\nI nominate Tobias for PTL of Gnocchi. Any objections? ;)\n\n \u003e I cannot state that enough: if you want to move a project, come to\n \u003e it, gain the trust of the maintainers, get (a copy of) the keys,\n \u003e and steer it. When old grumpy maintainers get annoyed with young\n \u003e and fresh new maintainers doing much of the work, they just give\n \u003e more and more way. :)\n \u003e \n \u003e (I don\u0027t think you were around when the Ceilometer team started,\n \u003e but you\u0027d be surprised by the diversity and the number of core\n \u003e reviewers we onboarded compared to the size of other OpenStack\n \u003e projects.)\n\nI was sort of around Ceilometer during Kilo/Liberty, but my contributions never made it upstream. My background was always writing custom pollsters and writing billing software based on data from Ceilometer. So I know about the project and was watching a lot of the work done on it, but never really got too involved upstream because of other priorities. We probably chatted on IRC at some stage about some pollster work I was doing.\n\n \u003e If you don\u0027t want to play that game for whatever reason or disagree\n \u003e with the maintainers, as many stated, go ahead and fork the\n \u003e project. I think it\u0027d be hard to justify a fork knowing that you\n \u003e didn\u0027t even try to send a single line of code to the project in\n \u003e your case — but it\u0027d not be my problem to justify such a decision.\n \u003e \n \u003e This is how open-source works — outside of the OpenStack bubble. :)\n\nThe sole reason we are in this mess is because if felt like patches and issues weren\u0027t getting looked at. Now they are, so lets maybe pivot to working on getting his fixes to CI done, and some bug fixes in. Then we can decide what to do.\n\nAnd forking a project when no existing devs seem to be active is very opensource too! :P\n\n\nAlright, how about this for a plan for Gnocchi:\n1. get CI passing on Github (with myself and Tobias helping)\n2. work on fixing some known bugs\n3. have a follow up discussion about its future, who its maintainers will be, and where it will live\n\nTelemetry in OS meanwhile needs to decide collectively what we want to do with our time series data (stick with Gnocchi, or move to something else).\n\nIf we do decide to move to something else, then I don\u0027t know what the future looks like for Gnocchi since OpenStack seems like the only people consuming it. It is probably going to be on life support until we make the switch/migration and fully deprecate the Gnocchi integrations. But again, given we\u0027d be the only people using it, it still does make sense to have OS people maintaining it.\n\nIf we do decide to stay with Gnocchi long term, then OpenDev is definitely the place for it given OpenStack is the only consumer, and that it was built to fit within OpenStack originally.\n\nI actually don\u0027t know what we\u0027d do. MongoDB is a huge NOPE, because that was a nightmare, but given InfluxDB enterprise is needed for clustering, that kinda looks like a bad option. TimescaleDB maybe works since it\u0027s based on PostgreSQL, but tuned for timeseries, although clent support looks like raw PostgreSQL. WhisperDB (graphite) isn\u0027t suited for our needs, nor is Prometheus really. We may well find that Gnocchi is actually still the best option for OpenStack despite not wanting to maintain a TSDB.\n\nI\u0027m parking this patch for now with a -1 workflow, but... I think eventually we\u0027ll end up redoing it, although hopefully as \u0027Gnocchi\u0027 rather than Farfalle or some other name like Kopytka.","accounts_in_message":[],"_revision_number":5},{"id":"000053c8bf8e1d0539d277c52b389467364c7095","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-09-10 11:35:15.000000000","message":"Patch Set 5:\n\n\u003e \u003e I\u0027m certain that if we step up, get CI working, make sure pending\n \u003e \u003e PR\u0027s is fine and can be merged, they will get merged (given they\n \u003e \u003e follow policy). If we get some people working on it more actively\n \u003e \u003e I\u0027m sure those persons will get merge rights.\n \u003e \n \u003e Yes, and if that doesn\u0027t happen, the thing to do then is do an\n \u003e actual fork. Same name, on opendev (but not OpenStack), new CI,\n \u003e that acknowledges that it is a fork and the reasons why. If you\n \u003e succeed you will no longer be the fork.\n \u003e \n \u003e All this noise about naming, governance and association with\n \u003e OpenStack is secondary to the real needs. Just do open source.\n \u003e Don\u0027t do this bureaucratic encumbered thing that OpenStack has\n \u003e normalized.\n\nThis was exactly my point. Totally agreed there.","accounts_in_message":[],"_revision_number":5},{"id":"97cbf323fb7c6f2e6f601a102885cf18c7c48cc3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2021-11-13 19:30:08.000000000","message":"Abandoned\n\nFeel free to restore if this is still needed","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"cb7537523f0b17012887bbd615bc38d0e1510015","revisions":{"c5a030d4709a0440acf7014343e75f9297e72c56":{"kind":"REWORK","_number":1,"created":"2020-08-04 02:51:11.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/92/744592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/744592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/744592/1"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-04 02:50:14.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-04 02:50:14.000000000","tz":720},"subject":"Return gnocchi back to openstack","message":"Return gnocchi back to openstack\n\nGiven that the codebase is unsupported elsewhere,\nwe are bringing it back in house to OpenStack as\nit is needed for telemetry and we have to support it.\n\nChange-Id: Ie1b916725c50c3f06a7d8561886da37839864959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5a030d4709a0440acf7014343e75f9297e72c56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c5a030d4709a0440acf7014343e75f9297e72c56"}]},"branch":"refs/heads/master"},"37415003d94f908fe1944481023160159cacb0aa":{"kind":"REWORK","_number":2,"created":"2020-08-28 00:06:30.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/92/744592/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/744592/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/744592/2"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-04 02:50:14.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-28 00:06:14.000000000","tz":720},"subject":"Fork Gnocchi back to openstack as Farfalle","message":"Fork Gnocchi back to openstack as Farfalle\n\nGnocchi is currently unmaintained:\nhttps://github.com/gnocchixyz/gnocchi/issues/1049\n\nPull requests to the Github repo to fix issues have been\nleft unreviewed and unmerged and there seems to be a lack\nof activity from the current devs behind the project.\n\nAn attempt was made to bring the project back over to OpenStack\nunder the same name where we could control governence for the\nproject and ideally still include existing contributors, but\ndespite no real activity from current project maintainers\nthey appear unwilling to relinquish control.\n\nAs such we are now proposing a fork back under a new namespace\n\u0027openstack-telemetry\u0027 where we can bring the project back to\nactive maintainence while we decide its future within OpenStack.\n\nGnocchi currently is the defacto offical Telemetry API as used by\nAodh, Cloudkitty, and relied on by thirdparty application around\nOpenStack as well. As such we do need to bring it back somewhat\nand maintain it.\n\nChange-Id: Ie1b916725c50c3f06a7d8561886da37839864959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37415003d94f908fe1944481023160159cacb0aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/37415003d94f908fe1944481023160159cacb0aa"}]},"branch":"refs/heads/master"},"eca389d565bfc660d6431edd2d89470f93d474af":{"kind":"REWORK","_number":3,"created":"2020-08-28 12:06:38.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/92/744592/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/744592/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/744592/3"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-04 02:50:14.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-28 12:05:09.000000000","tz":720},"subject":"Fork Gnocchi back to openstack as Farfalle","message":"Fork Gnocchi back to openstack as Farfalle\n\nGnocchi is currently unmaintained:\nhttps://github.com/gnocchixyz/gnocchi/issues/1049\n\nPull requests to the Github repo to fix issues have been\nleft unreviewed and unmerged and there seems to be a lack\nof activity from the current devs behind the project.\n\nAn attempt was made to bring the project back over to OpenStack\nunder the same name where we could control governence for the\nproject and ideally still include existing contributors, but\ndespite no real activity from current project maintainers\nthey appear unwilling to relinquish control.\n\nAs such we are now proposing a fork back under a new namespace\n\u0027farfalle\u0027 where we can bring the project back to active\nmaintainence while we decide its future within OpenStack.\n\nGnocchi currently is the defacto offical Telemetry API as used by\nAodh, Cloudkitty, and relied on by thirdparty applications around\nOpenStack as well. As such we do need to bring it back somewhat\nand maintain it.\n\nChange-Id: Ie1b916725c50c3f06a7d8561886da37839864959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eca389d565bfc660d6431edd2d89470f93d474af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eca389d565bfc660d6431edd2d89470f93d474af"}]},"branch":"refs/heads/master"},"0c05ac3219fd5ccb39a3d9b6b192fc1dd03941e2":{"kind":"REWORK","_number":4,"created":"2020-08-28 13:22:21.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/92/744592/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/744592/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/744592/4"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-04 02:50:14.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-28 13:22:09.000000000","tz":720},"subject":"Fork Gnocchi back to openstack as Farfalle","message":"Fork Gnocchi back to openstack as Farfalle\n\nGnocchi is currently unmaintained:\nhttps://github.com/gnocchixyz/gnocchi/issues/1049\n\nPull requests to the Github repo to fix issues have been\nleft unreviewed and unmerged and there seems to be a lack\nof activity from the current devs behind the project.\n\nAn attempt was made to bring the project back over to OpenStack\nunder the same name where we could control governence for the\nproject and ideally still include existing contributors, but\ndespite no real activity from current project maintainers\nthey appear unwilling to relinquish control.\n\nAs such we are now proposing a fork back under a new namespace\n\u0027farfalle\u0027 where we can bring the project back to active\nmaintainence while we decide its future within OpenStack.\n\nGnocchi currently is the defacto offical Telemetry API as used by\nAodh, Cloudkitty, and relied on by thirdparty applications around\nOpenStack as well. As such we do need to bring it back somewhat\nand maintain it.\n\nChange-Id: Ie1b916725c50c3f06a7d8561886da37839864959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c05ac3219fd5ccb39a3d9b6b192fc1dd03941e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0c05ac3219fd5ccb39a3d9b6b192fc1dd03941e2"}]},"branch":"refs/heads/master"},"cb7537523f0b17012887bbd615bc38d0e1510015":{"kind":"REWORK","_number":5,"created":"2020-08-28 14:02:44.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/92/744592/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/744592/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/744592/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/744592/5"}}},"commit":{"parents":[{"commit":"ec46579db40f29f96b2df9cabe987b96942ed281","subject":"Merge \"Remove os_congress gating\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ec46579db40f29f96b2df9cabe987b96942ed281"}]}],"author":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-28 14:02:28.000000000","tz":720},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-08-28 14:02:28.000000000","tz":720},"subject":"Fork Gnocchi back to openstack as Farfalle","message":"Fork Gnocchi back to openstack as Farfalle\n\nGnocchi is currently unmaintained:\nhttps://github.com/gnocchixyz/gnocchi/issues/1049\n\nPull requests to the Github repo to fix issues have been\nleft unreviewed and unmerged and there seems to be a lack\nof activity from the current devs behind the project.\n\nAn attempt was made to bring the project back over to OpenStack\nunder the same name where we could control governence for the\nproject and ideally still include existing contributors, but\ndespite no real activity from current project maintainers\nthey appear unwilling to relinquish control.\n\nAs such we are now proposing a fork back under a new namespace\n\u0027farfalle\u0027 where we can bring the project back to active\nmaintainence while we decide its future within OpenStack.\n\nGnocchi currently is the defacto offical Telemetry API as used by\nAodh, Cloudkitty, and relied on by thirdparty applications around\nOpenStack as well. As such we do need to bring it back somewhat\nand maintain it.\n\nChange-Id: Ie1b916725c50c3f06a7d8561886da37839864959\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb7537523f0b17012887bbd615bc38d0e1510015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb7537523f0b17012887bbd615bc38d0e1510015"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
