)]}'
{"id":"openstack%2Fproject-config~744890","triplet_id":"openstack%2Fproject-config~master~If4cf74bd5bf30a5436791fb4ebcf2a35006dbc8f","project":"openstack/project-config","branch":"master","topic":"keystone-kerberos-ceph-iscsi-github-mirror","hashtags":[],"change_id":"If4cf74bd5bf30a5436791fb4ebcf2a35006dbc8f","subject":"Mirror keystone-kerberos and ceph-iscsi charms to GitHub","status":"MERGED","created":"2020-08-05 11:54:53.000000000","updated":"2020-08-06 13:34:27.000000000","submitted":"2020-08-06 13:34:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"744890-1596720867607-c4c333d3","meta_rev_id":"c603a9f73bb019e5a7df30017878592c60e796ad","_number":744890,"virtual_id_number":744890,"owner":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 13:34:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-06 08:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-06 13:22:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-06 13:22:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-06 08:16:03.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-06 13:22:13.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-08-06 13:34:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4af5fb5048e43d723aef1499877407c750ff77f6","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2020-08-05 11:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"564e0f4ac7c742cae0c25756300b4415fb705983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 12:09:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/48cd19df9df947b8918a5b964cc1160a : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"03bf1ad5a6b925be40742c569e5ed1bef3717e8d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-05 12:18:01.000000000","message":"Patch Set 1:\n\nwant to add these repos as well: https://review.opendev.org/744479","accounts_in_message":[],"_revision_number":1},{"id":"fea0c3b9490b4d8b071cf0510e40da7ab37d1b92","author":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"date":"2020-08-05 13:29:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0ef1babb616f741066a1e527a680010c0ee6cd94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-05 13:45:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7b12575491cd4dc0881973bf5d1afaff : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"758f3cc27b66900d94d41811596a6ea183c2fb7c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-06 08:16:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f7ebeffed17688a8430c492d745c2c72d0ff564e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-08-06 13:22:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d9d298964b9a7d5d49f89b30abd0696195c0f764","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 13:22:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c48dffe722df91d0e0a2eb7ab5bffc8ba2471142","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 13:34:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e336ef3d6118e615674f53adda0250f8a3ec3f95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 13:34:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5f6b950040fe4fe79484c0b773430c7e : SUCCESS in 6m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b31161cc4c3d0de625e33fd9b5e904e90ab2745d","revisions":{"e34fb13b28379d24a467283fe8b88976c872bd9e":{"kind":"REWORK","_number":1,"created":"2020-08-05 11:54:53.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/90/744890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/744890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/744890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/744890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/744890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/744890/1"}}},"commit":{"parents":[{"commit":"8ea90306115c5cba60df25039b36287d9f5d21f6","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ea90306115c5cba60df25039b36287d9f5d21f6"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-08-05 11:53:54.000000000","tz":120},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-08-05 11:53:54.000000000","tz":120},"subject":"Mirror keystone-kerberos and ceph-iscsi charms to GitHub","message":"Mirror keystone-kerberos and ceph-iscsi charms to GitHub\n\nChange-Id: If4cf74bd5bf30a5436791fb4ebcf2a35006dbc8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e34fb13b28379d24a467283fe8b88976c872bd9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e34fb13b28379d24a467283fe8b88976c872bd9e"}]},"branch":"refs/heads/master"},"b31161cc4c3d0de625e33fd9b5e904e90ab2745d":{"kind":"REWORK","_number":2,"created":"2020-08-05 13:29:49.000000000","uploader":{"_account_id":31289,"name":"Aurelien Lourot","email":"aurelien.lourot@gmail.com","username":"lourot"},"ref":"refs/changes/90/744890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/744890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/744890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/744890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/744890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/744890/2"}}},"commit":{"parents":[{"commit":"8ea90306115c5cba60df25039b36287d9f5d21f6","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ea90306115c5cba60df25039b36287d9f5d21f6"}]}],"author":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-08-05 11:53:54.000000000","tz":120},"committer":{"name":"Aurelien Lourot","email":"aurelien.lourot@canonical.com","date":"2020-08-05 13:29:30.000000000","tz":120},"subject":"Mirror keystone-kerberos and ceph-iscsi charms to GitHub","message":"Mirror keystone-kerberos and ceph-iscsi charms to GitHub\n\nMirror also corresponding interfaces.\n\nChange-Id: If4cf74bd5bf30a5436791fb4ebcf2a35006dbc8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b31161cc4c3d0de625e33fd9b5e904e90ab2745d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b31161cc4c3d0de625e33fd9b5e904e90ab2745d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
