)]}'
{"id":"openstack%2Fproject-config~745990","triplet_id":"openstack%2Fproject-config~master~Ie82505632bbfe9ddfd4478e797f20f506ad260fc","project":"openstack/project-config","branch":"master","topic":"pyca-zuul","hashtags":[],"change_id":"Ie82505632bbfe9ddfd4478e797f20f506ad260fc","subject":"A pyca/cryptography to Zuul tenant","status":"ABANDONED","created":"2020-08-13 02:10:09.000000000","updated":"2020-08-13 07:44:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"db1ec74ffd58b60c9a3d5339213704d54d3a3e3d","_number":745990,"virtual_id_number":745990,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-13 02:19:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-13 07:20:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 02:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-13 07:20:44.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"84dc7e883c5388c877ca3008edb5bfcae40a12b1","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-13 02:10:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3580f05b63d7e2ecde3a33a5074d360c07d14f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 02:19:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d986c263406c4f82a25ae12cab19902c : SUCCESS in 5m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a8dfa393146e4fb9b76af9760a5724ed : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f452251cbbe65939fae0f4169b0eb26a965d5f95","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-13 07:20:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c19de428ccb5ba2a2fd1a8342963f2f2c27631c2","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-13 07:44:51.000000000","message":"Abandoned\n\nthis needs a special build container, it\u0027s not really useful","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"391e3b7287cedd409f99f39b26bf7a2c59dd4718","revisions":{"391e3b7287cedd409f99f39b26bf7a2c59dd4718":{"kind":"REWORK","_number":1,"created":"2020-08-13 02:10:09.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/90/745990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/90/745990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/745990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/745990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/90/745990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/90/745990/1"}}},"commit":{"parents":[{"commit":"e4120f97907a4e1cc6d9e8e6f29dfd4b75b5cf44","subject":"revert https://review.opendev.org/739605","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e4120f97907a4e1cc6d9e8e6f29dfd4b75b5cf44"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-08-13 02:08:22.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-08-13 02:08:22.000000000","tz":600},"subject":"A pyca/cryptography to Zuul tenant","message":"A pyca/cryptography to Zuul tenant\n\nThis is a good library to test building manylinux wheels with, so it\nwould be handy for zuul to clone it for zuul-jobs testing.\n\nChange-Id: Ie82505632bbfe9ddfd4478e797f20f506ad260fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/391e3b7287cedd409f99f39b26bf7a2c59dd4718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/391e3b7287cedd409f99f39b26bf7a2c59dd4718"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
