)]}'
{"id":"openstack%2Fproject-config~746407","triplet_id":"openstack%2Fproject-config~master~I22693569737365f4227a97c70b183167b51418c4","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I22693569737365f4227a97c70b183167b51418c4","subject":"Add vexxhost/tempest-pushgateway","status":"MERGED","created":"2020-08-16 15:26:19.000000000","updated":"2020-08-16 19:49:07.000000000","submitted":"2020-08-16 19:15:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746407-1597605315639-bd91aa04","meta_rev_id":"56409d84ed27540f2558db554b6c2528b7ab17a0","_number":746407,"virtual_id_number":746407,"owner":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-16 19:15:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-16 19:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-08-16 15:42:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-08-16 19:07:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-16 15:42:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-08-16 19:07:55.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-16 19:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"58377dd6bc595d9a56b9efbb80a4a1780e170795","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-08-16 15:26:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9850c0d657ed45d24b23107a17acc84fe05b4798","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-16 15:30:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e8aad7e3099164a7106e29fb0cb9fe2a7764097a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-16 15:30:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"394281b165ebce51f4519af0dc1d2aa216f10149","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-08-16 15:33:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f7aa1db1a3bf6f985559148cd035a5c047ea56c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 15:39:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bae19d313e4f4926bb161572d8667463 : SUCCESS in 5m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/593f0ecf26e7476e85817bc28b657956 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f69ae3d7c18780d489d032257f310fb846009dad","author":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"date":"2020-08-16 15:40:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"350fcad6c2f5ed14acf13d6540e6c2539985ac06","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-08-16 15:42:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d9ca256c1097e9a63794d68c8fc242b30c998dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 15:46:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f4fb8e0bc02546748ea552fa34193b16 : SUCCESS in 5m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/84f9c3f9ff0347e7b81173e1fe581b81 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":3},{"id":"3af2b72360ee9d641e89eab224e9e3d5607714e4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-16 19:07:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"04e69420e92f2d40fb772bfd3d94e846348ed7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 19:08:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2735f7dcc01eb440d4920a48fe7ef604a35ba3ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 19:15:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"294c785346840cd0818158dcd40fee3124e47963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 19:15:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/85c75de0785842d4b60c53204ec703e3 : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7d918bc5935f41688a090c8bd20071fb : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"56409d84ed27540f2558db554b6c2528b7ab17a0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-16 19:49:07.000000000","message":"Patch Set 3:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/8449f8d4b67742cb82f9ca90fe2bf4bf : SUCCESS in 2m 50s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/49b7aa3e0dc4459cabb28cab662fae57 : TIMED_OUT in 30m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b3d66849d0f5e53cb0988b9ac031a28aea542e14","revisions":{"557702364167516013a1c77bb8f56b0d0b40a39f":{"kind":"REWORK","_number":1,"created":"2020-08-16 15:26:19.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/07/746407/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/746407/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/746407/1"}}},"commit":{"parents":[{"commit":"39b1e8d55a21998f56746012b10d968955329140","subject":"Merge \"Create pyca/infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39b1e8d55a21998f56746012b10d968955329140"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-08-16 15:25:08.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-08-16 15:26:14.000000000","tz":180},"subject":"Add vexxhost/tempest-pushgatway","message":"Add vexxhost/tempest-pushgatway\n\nChange-Id: I22693569737365f4227a97c70b183167b51418c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/557702364167516013a1c77bb8f56b0d0b40a39f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/557702364167516013a1c77bb8f56b0d0b40a39f"}]},"branch":"refs/heads/master"},"f22c4e7e1c29c8b92914252f80f803e32cc0d1f6":{"kind":"REWORK","_number":2,"created":"2020-08-16 15:33:46.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/07/746407/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/746407/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/746407/2"}}},"commit":{"parents":[{"commit":"39b1e8d55a21998f56746012b10d968955329140","subject":"Merge \"Create pyca/infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39b1e8d55a21998f56746012b10d968955329140"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-08-16 15:25:08.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-08-16 15:33:26.000000000","tz":180},"subject":"Add vexxhost/tempest-pushgateway","message":"Add vexxhost/tempest-pushgateway\n\nChange-Id: I22693569737365f4227a97c70b183167b51418c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f22c4e7e1c29c8b92914252f80f803e32cc0d1f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f22c4e7e1c29c8b92914252f80f803e32cc0d1f6"}]},"branch":"refs/heads/master"},"b3d66849d0f5e53cb0988b9ac031a28aea542e14":{"kind":"REWORK","_number":3,"created":"2020-08-16 15:40:22.000000000","uploader":{"_account_id":31746,"name":"Oleksandr Kozachenko","email":"okozachenko1203@gmail.com","username":"okozachenko"},"ref":"refs/changes/07/746407/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/07/746407/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/07/746407/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/07/746407/3"}}},"commit":{"parents":[{"commit":"39b1e8d55a21998f56746012b10d968955329140","subject":"Merge \"Create pyca/infra\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/39b1e8d55a21998f56746012b10d968955329140"}]}],"author":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-08-16 15:25:08.000000000","tz":180},"committer":{"name":"okozachenko","email":"okozachenko@vexxhost.com","date":"2020-08-16 15:40:05.000000000","tz":180},"subject":"Add vexxhost/tempest-pushgateway","message":"Add vexxhost/tempest-pushgateway\n\nChange-Id: I22693569737365f4227a97c70b183167b51418c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3d66849d0f5e53cb0988b9ac031a28aea542e14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b3d66849d0f5e53cb0988b9ac031a28aea542e14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
