)]}'
{"id":"openstack%2Fproject-config~746759","triplet_id":"openstack%2Fproject-config~master~Ib6dadc423a0b794bcf85b70add1eeae8974aad79","project":"openstack/project-config","branch":"master","topic":"puppet-tripleo","hashtags":[],"change_id":"Ib6dadc423a0b794bcf85b70add1eeae8974aad79","subject":"Re-introduce puppet-tripleo-core group","status":"MERGED","created":"2020-08-18 16:21:14.000000000","updated":"2020-08-20 21:06:56.000000000","submitted":"2020-08-20 21:00:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746759-1597957223667-0adfb4b4","meta_rev_id":"a5feaed48fe9129cc430e74b90dcf008a19ab1f5","_number":746759,"virtual_id_number":746759,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-20 21:00:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-20 20:52:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-08-20 06:54:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-20 20:52:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-18 16:21:58.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-18 16:22:58.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2020-08-19 15:18:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-20 06:54:42.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-20 20:52:34.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-08-20 21:00:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09b886d698522a4bc19be4b0537a51c81825e47f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-18 16:21:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab8ccf233358d67671734dee6277270a6ac0a05f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-18 16:21:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6d221c80b17189308daa0332d051e8d018a9549c","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2020-08-18 16:22:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cf58cbe02e57cadf0c27d375641344ae01a1330b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 16:28:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3dbafc5bcb1c472c8e5909a46587e88e : FAILURE in 6m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2c538115bb714f07981d06fc123901af : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"db77d77d80f00571dea869d8c7029d4117611ff1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-18 21:04:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d5505f05d047e1f554240b7cdf15fb3aa7daa02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-18 21:15:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d951591087144e1a1a76eb6b54b5906 : FAILURE in 5m 00s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fab4e5f7f97946e09f770a8341168397 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f4abe6312a012e0fa47f24a31f395179974a251b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 10:33:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"85a639d571e540a12d31798b6780bdbf60e80020","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 13:04:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7612587bbc5432169e8bdb518e1212df2d81a31","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 13:17:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ffd539831cfb537637e936972c8fd0c6e897ae2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 13:30:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1f553fac0f884aaf8d0efb2173c6b21c : FAILURE in 5m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/64558cb26e784efd99d917f2816698dd : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"ef4c3699f08c4812d8da6907d70cf3f82e2f2106","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 13:53:12.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"f6fc86dc39ba10b62602906267215ade3672c5b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 14:11:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/44c064578c8b4adca88b09f17b56c47b : SUCCESS in 6m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/79463aaf76a2400097a36325d91ee302 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"8a95cd27c438c22284951c5546b987477628b4aa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 15:18:58.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"8c60139febc90c358435ee9c9354a77de83157c6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-19 19:18:53.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nThis includes a change that is not documented.","accounts_in_message":[],"_revision_number":4},{"id":"4fd80d24a39b9d9dce2eecd465541a5ec88a4a0f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 21:14:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dae6c4d70b234a6a1eb2d0ddcad02bd40fd68fec","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 21:14:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f0e208c308ffc35b6a3e257796cac54499353356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 21:28:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1748bd9df6a44d489c52cac30fed5b85 : SUCCESS in 5m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0d28b7db81e64e88b23a09b0777613cb : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":5},{"id":"b1f5bac94721a7170a83eb92147a9ebe6105aa7f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-20 06:54:42.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d865ade6ea25d047e16f99f5b473684b7a02226b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-20 20:52:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"b623eb126dffbad35b0bd8a1078e14b5d45418e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 20:52:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"385a1a7d99c8c2525598607f970d5a86b7393898","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:00:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b64dab62f99f47cab13067d4e958ba9d : SUCCESS in 6m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/068617493b8043ffbe1e912d3797092f : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":5},{"id":"9c126f1a9b2070d54ccbcd734cfc62f7fcc91459","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:00:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8d5837ffece10e9c48ec7aa37c318a23037ce15d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 21:06:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/492ce5c3c5f34d09a9cec5f9ce0f6dd0 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8e169c45edcb23ab0a26e5b78b0d515396eb48fb","revisions":{"e8e0509d148deec6e37ffb3be3550d4cb770424c":{"kind":"REWORK","_number":1,"created":"2020-08-18 16:21:14.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/59/746759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/746759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/746759/1"}}},"commit":{"parents":[{"commit":"cad6162de17d25c8129d142d80e31da5bb828d9e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cad6162de17d25c8129d142d80e31da5bb828d9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 16:20:01.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 16:20:44.000000000","tz":-240},"subject":"Re-introduce puppet-tripleo-core group","message":"Re-introduce puppet-tripleo-core group\n\nCreating new ACLs for puppet-tripleo project, so we can have a group of\ncores dedicated to that repo, and include the tripleo-core group\nmaintainer of that project as well.\n\nIt\u0027ll allow the TripleO project to have finer gerrit permissions for\nfolks who are only core on puppet-tripleo.\n\nChange-Id: Ib6dadc423a0b794bcf85b70add1eeae8974aad79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8e0509d148deec6e37ffb3be3550d4cb770424c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e8e0509d148deec6e37ffb3be3550d4cb770424c"}]},"branch":"refs/heads/master"},"0838335ef0f6d6dcbd7d4ba21c69988963a7ecb0":{"kind":"REWORK","_number":2,"created":"2020-08-18 21:04:21.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/59/746759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/746759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/746759/2"}}},"commit":{"parents":[{"commit":"cad6162de17d25c8129d142d80e31da5bb828d9e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cad6162de17d25c8129d142d80e31da5bb828d9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 16:20:01.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 21:04:15.000000000","tz":-240},"subject":"Re-introduce puppet-tripleo-core group","message":"Re-introduce puppet-tripleo-core group\n\nCreating new ACLs for puppet-tripleo project, so we can have a group of\ncores dedicated to that repo, and include the tripleo-core group\nmaintainer of that project as well.\n\nIt\u0027ll allow the TripleO project to have finer gerrit permissions for\nfolks who are only core on puppet-tripleo.\n\nChange-Id: Ib6dadc423a0b794bcf85b70add1eeae8974aad79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0838335ef0f6d6dcbd7d4ba21c69988963a7ecb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0838335ef0f6d6dcbd7d4ba21c69988963a7ecb0"}]},"branch":"refs/heads/master"},"825943eeeda659865ac7d79a3c27c8f1bdf1f84f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-08-19 13:17:52.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/59/746759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/746759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/746759/3"}}},"commit":{"parents":[{"commit":"fd69d3d33f3bc2fd2f7906d01a76a5a0bad4018b","subject":"Disable E208 for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd69d3d33f3bc2fd2f7906d01a76a5a0bad4018b"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 16:20:01.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 13:17:52.000000000","tz":0},"subject":"Re-introduce puppet-tripleo-core group","message":"Re-introduce puppet-tripleo-core group\n\nCreating new ACLs for puppet-tripleo project, so we can have a group of\ncores dedicated to that repo, and include the tripleo-core group\nmaintainer of that project as well.\n\nIt\u0027ll allow the TripleO project to have finer gerrit permissions for\nfolks who are only core on puppet-tripleo.\n\nChange-Id: Ib6dadc423a0b794bcf85b70add1eeae8974aad79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/825943eeeda659865ac7d79a3c27c8f1bdf1f84f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/825943eeeda659865ac7d79a3c27c8f1bdf1f84f"}]},"branch":"refs/heads/master"},"9a38ff8cae4d0070557225b1bdb7cf810907ac60":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-08-19 13:53:12.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/59/746759/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/746759/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/746759/4"}}},"commit":{"parents":[{"commit":"d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f","subject":"Disable E106 \u0026 E208 for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 16:20:01.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 13:53:12.000000000","tz":0},"subject":"Re-introduce puppet-tripleo-core group","message":"Re-introduce puppet-tripleo-core group\n\nCreating new ACLs for puppet-tripleo project, so we can have a group of\ncores dedicated to that repo, and include the tripleo-core group\nmaintainer of that project as well.\n\nIt\u0027ll allow the TripleO project to have finer gerrit permissions for\nfolks who are only core on puppet-tripleo.\n\nChange-Id: Ib6dadc423a0b794bcf85b70add1eeae8974aad79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a38ff8cae4d0070557225b1bdb7cf810907ac60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9a38ff8cae4d0070557225b1bdb7cf810907ac60"}]},"branch":"refs/heads/master"},"8e169c45edcb23ab0a26e5b78b0d515396eb48fb":{"kind":"REWORK","_number":5,"created":"2020-08-19 21:14:30.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/59/746759/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/59/746759/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/59/746759/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/59/746759/5"}}},"commit":{"parents":[{"commit":"d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f","subject":"Disable E106 \u0026 E208 for now","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-18 16:20:01.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 21:14:17.000000000","tz":-240},"subject":"Re-introduce puppet-tripleo-core group","message":"Re-introduce puppet-tripleo-core group\n\nCreating new ACLs for puppet-tripleo project, so we can have a group of\ncores dedicated to that repo, and include the tripleo-core group\nmaintainer of that project as well.\n\nIt\u0027ll allow the TripleO project to have finer gerrit permissions for\nfolks who are only core on puppet-tripleo.\n\nChange-Id: Ib6dadc423a0b794bcf85b70add1eeae8974aad79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e169c45edcb23ab0a26e5b78b0d515396eb48fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8e169c45edcb23ab0a26e5b78b0d515396eb48fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
