)]}'
{"id":"openstack%2Fproject-config~746920","triplet_id":"openstack%2Fproject-config~master~I0fdd4a79b886d1c4875e748b08f99da809d9d767","project":"openstack/project-config","branch":"master","topic":"ansible-lint-E208","hashtags":[],"change_id":"I0fdd4a79b886d1c4875e748b08f99da809d9d767","subject":"Disable E106 \u0026 E208 for now","status":"MERGED","created":"2020-08-19 13:17:30.000000000","updated":"2020-08-19 15:28:17.000000000","submitted":"2020-08-19 15:28:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746920-1597850897055-724cfc27","meta_rev_id":"c34d489e3505025177d7ac86a40aad4b87993c0d","_number":746920,"virtual_id_number":746920,"owner":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-19 15:28:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-08-19 13:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-08-19 15:19:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-19 15:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-19 15:18:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-19 13:56:57.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-08-19 15:18:18.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-08-19 15:19:31.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-08-19 15:28:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"076f5c4380645436efdaf8f98079258edebb134f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 13:17:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44ee92c51a6fcdc2aab1643d8564014e86075e2d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-19 13:23:36.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":1},{"id":"8b6361fdf2e853fa71934c98eff39e13d25fd3a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 13:30:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ea3482152345415f93665a3b9d0dde3b : FAILURE in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9f563e0fdb8ab85e968ec028a68b81bb77b4144b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-19 13:52:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"098bd75b63a08697757ce755d7f8e78aec4b5874","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-19 13:56:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0077b75c5286620770058bd07aa4a4d6ada6cbab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 14:04:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/63b4f6d505fd43e0b55fe9d635db37ec : SUCCESS in 8m 19s","accounts_in_message":[],"_revision_number":2},{"id":"47ce3ad600ac355cd622834e4de601fd3ef7cbd5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-08-19 15:18:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cde644945970b83cb1b2cb50a66bf0fa168b20c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 15:18:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"220c718f66ad1b5e383f0b6a8108c07c6b352072","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-19 15:19:31.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"075070d6ddcef40ef1f5a5bf78ccc70f421e1edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 15:28:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/caf36540b8c24d3d936f1d14de7c24bb : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":2},{"id":"becf18b49e6221dddb33cf8f01735357b405fe75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 15:28:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f","revisions":{"fd69d3d33f3bc2fd2f7906d01a76a5a0bad4018b":{"kind":"REWORK","_number":1,"created":"2020-08-19 13:17:30.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/20/746920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/746920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/746920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/746920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/746920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/746920/1"}}},"commit":{"parents":[{"commit":"cad6162de17d25c8129d142d80e31da5bb828d9e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cad6162de17d25c8129d142d80e31da5bb828d9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 13:16:39.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 13:16:39.000000000","tz":-240},"subject":"Disable E208 for now","message":"Disable E208 for now\n\nansible-lint throws an undocumented error \"E208 File permissions not\nmentioned\" that looks too aggressive, for example it complains IMHO\nwrongly about:\n\nroles/add-build-sshkey/tasks/remote-linux.yaml:1: [E208] File\npermissions not mentioned\nroles/add-build-sshkey/tasks/remote-linux.yaml:14: [E208] File\npermissions not mentioned\nroles/add-build-sshkey/tasks/remote-linux.yaml:20: [E208] File\npermissions not mentioned\nroles/add-build-sshkey/tasks/remote-linux.yaml:27: [E208] File\npermissions not mentioned\n\nDisable it for now until ansible-lint is fixed - and the real problems\nin zuul-jobs are fixed.\n\nhttps://github.com/ansible/ansible-lint/pull/949 fixes the false\npositives. Once that is merged and a new release is out, we should\nrevert and fix the real problems.\n\nInspired by I3e3978bbc9458ad79623d22969344fb46f5fee43\n\nChange-Id: I0fdd4a79b886d1c4875e748b08f99da809d9d767\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd69d3d33f3bc2fd2f7906d01a76a5a0bad4018b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fd69d3d33f3bc2fd2f7906d01a76a5a0bad4018b"}]},"branch":"refs/heads/master"},"d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f":{"kind":"REWORK","_number":2,"created":"2020-08-19 13:52:34.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/20/746920/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/20/746920/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/746920/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/746920/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/20/746920/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/20/746920/2"}}},"commit":{"parents":[{"commit":"cad6162de17d25c8129d142d80e31da5bb828d9e","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cad6162de17d25c8129d142d80e31da5bb828d9e"}]}],"author":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 13:16:39.000000000","tz":-240},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2020-08-19 13:51:10.000000000","tz":-240},"subject":"Disable E106 \u0026 E208 for now","message":"Disable E106 \u0026 E208 for now\n\nE106 rejects roles with \u0027-\u0027 in their names, which is important\nfor collections, but these aren\u0027t collections.\n\nAlso ansible-lint throws an undocumented error \"E208 File permissions not\nmentioned\" that looks too aggressive, for example it complains IMHO\nwrongly about:\n\nroles/add-build-sshkey/tasks/remote-linux.yaml:1: [E208] File\npermissions not mentioned\nroles/add-build-sshkey/tasks/remote-linux.yaml:14: [E208] File\npermissions not mentioned\nroles/add-build-sshkey/tasks/remote-linux.yaml:20: [E208] File\npermissions not mentioned\nroles/add-build-sshkey/tasks/remote-linux.yaml:27: [E208] File\npermissions not mentioned\n\nDisable it for now until ansible-lint is fixed - and the real problems\nin zuul-jobs are fixed.\n\nhttps://github.com/ansible/ansible-lint/pull/949 fixes the false\npositives. Once that is merged and a new release is out, we should\nrevert and fix the real problems.\n\nInspired by I3e3978bbc9458ad79623d22969344fb46f5fee43\nand I48f837e8d5bd7d0d9d708d728d93627deb4093cc\n\nChange-Id: I0fdd4a79b886d1c4875e748b08f99da809d9d767\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nCo-Authored-By: James E. Blair \u003cjeblair@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1d0f66734ad1c3c7e74e82d23d44a5f924a0d2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
