)]}'
{"id":"openstack%2Fproject-config~747312","triplet_id":"openstack%2Fproject-config~master~I94290ed8c37ff5b402f90e6eeec1e4fd34893c2a","project":"openstack/project-config","branch":"master","topic":"unpause-git-removal","hashtags":[],"change_id":"I94290ed8c37ff5b402f90e6eeec1e4fd34893c2a","subject":"Revert \"Pause all image builds\"","status":"MERGED","created":"2020-08-20 23:23:51.000000000","updated":"2020-08-21 00:52:49.000000000","submitted":"2020-08-21 00:40:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747312-1597970440875-14551777","meta_rev_id":"79af89feb1da0b49057be36e498c4c447dd1a8be","_number":747312,"virtual_id_number":747312,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 00:40:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-08-21 00:24:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 00:31:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 00:24:56.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-08-21 00:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ad0f66708af045881dc95b0ee6027b7b2b6582f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-20 23:23:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a96b166245513cdddad2d9a11d8284dae7bbe6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 23:38:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3e5bee604e9848a1a290d5482dbd64dd : SUCCESS in 7m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/05e874ce66054766bef0dc0ded3918a7 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"0c9b6aba125c34b1fb247dfb3fc9ecfc5573dd98","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-08-21 00:24:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d38e28ac34e987931c2a4ff3ef545200b42b1067","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-08-21 00:31:04.000000000","message":"Patch Set 1: Workflow+1\n\nI\u0027ve confirmed the builders have 3.2.1\n\n$ sudo docker exec -it nodepool-builder-compose_nodepool-builder_1 disk-image-create --version\n3.2.1","accounts_in_message":[],"_revision_number":1},{"id":"df9f64dfb45b506048772caf7c755931d2671121","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:31:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"813d8f2dc6d7bda31b67f9c67cca2ea1ac287013","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:40:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7d80077b2cc3439ca31ff873830ab75d : SUCCESS in 7m 18s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/3bbb8b9ef0114373a72e6a7d11f536f8 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2e8d4c4ba1f281899b4383b89e357ec30febcde9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:40:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e1874b45a3f43a0bc7010448a07b661ce24b8758","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 00:52:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/a42ab7a1953741098f8e829339fe3079 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"88a8bf3aa613120c31a2d6198686feb41bdc192e","revisions":{"88a8bf3aa613120c31a2d6198686feb41bdc192e":{"kind":"REWORK","_number":1,"created":"2020-08-20 23:23:51.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/12/747312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/747312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/747312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/747312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/747312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/747312/1"}}},"commit":{"parents":[{"commit":"3643fbbbcf39c490d7ae4cdc35cf6ee4dc6e482a","subject":"Merge \"Re-introduce puppet-tripleo-core group\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3643fbbbcf39c490d7ae4cdc35cf6ee4dc6e482a"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-08-20 23:22:55.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-08-20 23:23:29.000000000","tz":600},"subject":"Revert \"Pause all image builds\"","message":"Revert \"Pause all image builds\"\n\nThis reverts commit 37461abc86a972e5a16000970023a65f142d9603.\n\nThis should be fixed with dib 3.2.1 which has rolled out via\nhttps://review.opendev.org/#/c/747303/\n\nChange-Id: I94290ed8c37ff5b402f90e6eeec1e4fd34893c2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a8bf3aa613120c31a2d6198686feb41bdc192e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/88a8bf3aa613120c31a2d6198686feb41bdc192e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
