)]}'
{"id":"openstack%2Fproject-config~748304","triplet_id":"openstack%2Fproject-config~master~I13b8ff3a874db1856b8f10cb79eb52cb3d4d18ee","project":"openstack/project-config","branch":"master","hashtags":[],"change_id":"I13b8ff3a874db1856b8f10cb79eb52cb3d4d18ee","subject":"Add openstack/openstack-helm-deployments","status":"MERGED","created":"2020-08-26 20:33:16.000000000","updated":"2020-09-01 19:10:37.000000000","submitted":"2020-09-01 18:36:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748304-1598985377629-436c11b4","meta_rev_id":"6c55e3abd65184fa4c33c18b37a7dea2ee16aa34","_number":748304,"virtual_id_number":748304,"owner":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 18:36:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-09-01 14:47:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-09-01 18:24:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-09-01 18:24:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"value":0,"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 20:34:16.000000000","updated_by":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"reviewer":{"_account_id":22477,"name":"Matt McEuen","email":"matt.mceuen@att.com","username":"mattmceuen"},"state":"REVIEWER"},{"updated":"2020-09-01 14:47:29.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-01 18:24:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-09-01 18:36:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a6e53e1d0e03c7e3d86037f3b8769412583f1d4","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-08-26 20:33:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f7ab1a654e0914c5f543305501e352a5c3ab5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 20:57:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b99734af623642aba35817f5adc4dd7f : SUCCESS in 8m 49s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7e0b7652c6c94b0c8989a6ae38ffbe24 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"effb57ab3fb77db9bd507f0fa156fe00cf3a6d18","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-27 14:56:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c643b8c6e6c4bd2791979d9b8907b1f1e40b6fbd","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-08-27 14:58:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2232b6815abba68b828c859278e21ad069c8da9b","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-01 00:52:50.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f52683da747cdf164873dda78b8dfdca769fdf96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 01:00:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ae9d7990ed9e48afa2572244df9d9e57 : SUCCESS in 6m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/35c705d1f45d4335adc766a029a774a5 : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"702ddf8d859b870af634898d8c8b752b9dac0bfa","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-01 05:37:47.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9ba370dd6ba5b8a35dd52d9dbd2d16c13450beef","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-01 14:15:37.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"89eecf635cbd5bfa84dc3a3c85f5098815fd9ec7","author":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"date":"2020-09-01 14:16:08.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e0f845964bdba373a6b023d4710be39f89a081e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:38:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2b797b4bb0404df0a43ccfdf1975afdc : SUCCESS in 11m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/319c856e85d540dc85c9e12b59ef4df3 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"a50efdfd75e67b307e70497071b9a7d2f3890f51","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-01 14:47:29.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"58f9c1bc7b4524c36ec394c346ee5562fc043475","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-01 18:24:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9088f6b3679d8ecb6bc7468b80dde462c2b4a44d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:25:09.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d09029a839cb6620970953e5d71f072ae1605615","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:36:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"69111f01ffd27a64b8408601e0eb3ba43acc5607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 18:36:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f0c2210c75bf447a82238087ce2de6c0 : SUCCESS in 5m 52s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/425c3bdfe04b4526b370ef5c52b23ef7 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":3},{"id":"6c55e3abd65184fa4c33c18b37a7dea2ee16aa34","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 19:10:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c1bf10d9d494479caacb03607579c370 : SUCCESS in 2m 50s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/e372085296fc40d5868b9379ede431e4 : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"44fef44c26dfd4d83e132543d5fc80ff6b7c1463","revisions":{"ac0dd598a6554950e8edeeed29dd1b683a8c88b6":{"kind":"REWORK","_number":1,"created":"2020-08-26 20:33:16.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/04/748304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/748304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/748304/1"}}},"commit":{"parents":[{"commit":"25dacf963d147dbfccb5c5f03da4fb489242d56a","subject":"Merge \"Do not update upper constraints for intermediate branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25dacf963d147dbfccb5c5f03da4fb489242d56a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-08-26 20:31:32.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-08-26 20:31:32.000000000","tz":-300},"subject":"Add openstack/openstack-helm-releases","message":"Add openstack/openstack-helm-releases\n\nThis change adds a new repo under the openstack-helm project\nin order to provide basic deploymeny declarations for\nopenstack-helm charts.\n\nChange-Id: I13b8ff3a874db1856b8f10cb79eb52cb3d4d18ee\nNeeded-By: https://review.opendev.org/#/c/748302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac0dd598a6554950e8edeeed29dd1b683a8c88b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac0dd598a6554950e8edeeed29dd1b683a8c88b6"}]},"branch":"refs/heads/master"},"b126673dfe9c387344afe2e3d13a76110d8bed01":{"kind":"REWORK","_number":2,"created":"2020-09-01 00:52:50.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/04/748304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/748304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/748304/2"}}},"commit":{"parents":[{"commit":"25dacf963d147dbfccb5c5f03da4fb489242d56a","subject":"Merge \"Do not update upper constraints for intermediate branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25dacf963d147dbfccb5c5f03da4fb489242d56a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-08-26 20:31:32.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-09-01 00:52:47.000000000","tz":0},"subject":"Add openstack/openstack-helm-deployments","message":"Add openstack/openstack-helm-deployments\n\nThis change adds a new repo under the openstack-helm project\nin order to provide basic deployment declarations for\nopenstack-helm charts.\n\nChange-Id: I13b8ff3a874db1856b8f10cb79eb52cb3d4d18ee\nNeeded-By: https://review.opendev.org/#/c/748302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b126673dfe9c387344afe2e3d13a76110d8bed01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b126673dfe9c387344afe2e3d13a76110d8bed01"}]},"branch":"refs/heads/master"},"44fef44c26dfd4d83e132543d5fc80ff6b7c1463":{"kind":"REWORK","_number":3,"created":"2020-09-01 14:15:37.000000000","uploader":{"_account_id":21420,"name":"Gage Hugo","email":"gagehugo@gmail.com","username":"ghugo"},"ref":"refs/changes/04/748304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/748304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/748304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/748304/3"}}},"commit":{"parents":[{"commit":"25dacf963d147dbfccb5c5f03da4fb489242d56a","subject":"Merge \"Do not update upper constraints for intermediate branches\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/25dacf963d147dbfccb5c5f03da4fb489242d56a"}]}],"author":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-08-26 20:31:32.000000000","tz":-300},"committer":{"name":"Gage Hugo","email":"gagehugo@gmail.com","date":"2020-09-01 14:15:35.000000000","tz":0},"subject":"Add openstack/openstack-helm-deployments","message":"Add openstack/openstack-helm-deployments\n\nThis change adds a new repo under the openstack-helm project\nin order to provide basic deployment declarations for\nopenstack-helm charts.\n\nChange-Id: I13b8ff3a874db1856b8f10cb79eb52cb3d4d18ee\nNeeded-By: https://review.opendev.org/#/c/748302\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/44fef44c26dfd4d83e132543d5fc80ff6b7c1463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/44fef44c26dfd4d83e132543d5fc80ff6b7c1463"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
