)]}'
{"id":"openstack%2Fproject-config~748377","triplet_id":"openstack%2Fproject-config~master~I0f74f0000b6639be2be9e8537a2b4cacdfc21e97","project":"openstack/project-config","branch":"master","topic":"backport_candidates","hashtags":[],"change_id":"I0f74f0000b6639be2be9e8537a2b4cacdfc21e97","subject":"Add Backport-Candidate label for nova deliverables","status":"ABANDONED","created":"2020-08-27 08:49:44.000000000","updated":"2021-01-12 19:19:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"852919d5edc4ca2e93355a732b3758f036db882f","_number":748377,"virtual_id_number":748377,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2021-01-12 19:19:41.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-08-27 09:26:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-08-27 08:51:08.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-27 08:51:08.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-08-27 09:26:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-12 18:43:26.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"}],"messages":[{"id":"89087351f8e6ec261db551a0f36d790270edafba","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-27 08:49:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76a0d46cd84989b55601be46df453b792aefa6f1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-27 08:51:01.000000000","message":"Patch Set 1:\n\nNote that I have done this differently to how oslo do it, dropping the +2/-2 in favour of core-only +1/-1. Does that make sense?","accounts_in_message":[],"_revision_number":1},{"id":"2d0ba5c7fd2923842fc9063216f17931e0af695d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-27 09:02:06.000000000","message":"Patch Set 1:\n\nI wonder if we want to give stable-cores +1 backport-candidate rights too?","accounts_in_message":[],"_revision_number":1},{"id":"ead7afc32920a5df5300f6bc10b58e611d2ae1e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 09:26:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d288e48640eb4b459db601c2633f85bc : SUCCESS in 5m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/23c0425cd3fa426aaa6121048e7d3fed : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"83674cd22f3cb64fd26722886e11b7138465e119","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-01-12 18:43:26.000000000","message":"Patch Set 1:\n\nIs there still an interest in doing this? Happy to discuss further if so, though do also check out new Gerrit features since the upgrade, particularly please evaluate whether the hashtags feature would be more appropriate.","accounts_in_message":[],"_revision_number":1},{"id":"d9cbd3ee1e38bc0edc45a0582f3ed6b9ddbb0ad9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-12 19:19:41.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Is there still an interest in doing this? Happy to discuss further if so, though do also check out new Gerrit features since the upgrade, particularly please evaluate whether the hashtags feature would be more appropriate.\n\nProbably not. We agreed to do it but having used it in Oslo land, I think it\u0027s less useful than I thought previously. I\u0027ll look at those new features.","accounts_in_message":[],"_revision_number":1},{"id":"852919d5edc4ca2e93355a732b3758f036db882f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-01-12 19:19:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81bf17a31076bdbc9c35ec74d6129c62c354c57d","revisions":{"81bf17a31076bdbc9c35ec74d6129c62c354c57d":{"kind":"REWORK","_number":1,"created":"2020-08-27 08:49:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/77/748377/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/77/748377/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/748377/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/748377/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/77/748377/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/77/748377/1"}}},"commit":{"parents":[{"commit":"907bfdca3c5d683e02149d271699cfabbcca7fce","subject":"Merge \"Import netowrking-l2gw \u0026 networking-l2gw-tempest-plugin to x/\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/907bfdca3c5d683e02149d271699cfabbcca7fce"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-08-27 08:42:10.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-08-27 08:42:10.000000000","tz":60},"subject":"Add Backport-Candidate label for nova deliverables","message":"Add Backport-Candidate label for nova deliverables\n\nList of deliverables taken from [1].\n\n[1] https://releases.openstack.org/teams/nova.html\n\nChange-Id: I0f74f0000b6639be2be9e8537a2b4cacdfc21e97\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/81bf17a31076bdbc9c35ec74d6129c62c354c57d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/81bf17a31076bdbc9c35ec74d6129c62c354c57d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
