)]}'
{"id":"openstack%2Fproject-config~748683","triplet_id":"openstack%2Fproject-config~master~I46c63250c4081c9915dc83b61880ad88f63f8efc","project":"openstack/project-config","branch":"master","topic":"osa/os_senlin","hashtags":[],"change_id":"I46c63250c4081c9915dc83b61880ad88f63f8efc","subject":"Add openstack-ansible/os_senlin role","status":"MERGED","created":"2020-08-28 14:53:31.000000000","updated":"2020-09-01 15:12:48.000000000","submitted":"2020-09-01 14:43:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748683-1598971389582-9d651cf1","meta_rev_id":"9e920f7c622c0cf155ab5d94aa084d9a73dec328","_number":748683,"virtual_id_number":748683,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"date":"2020-09-01 06:08:15.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-01 14:43:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":2,"date":"2020-09-01 14:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2020-09-01 13:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-09-01 14:24:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 13:12:31.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-09-01 14:24:14.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-01 14:43:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099205d3536b3d3253a4690aa97fe299deb6ff4f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-28 14:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60f90c0fb015779a97a42d44112876de41016231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 15:02:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8a68d7db83b94d8bbb4b8081b205d860 : SUCCESS in 6m 10s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/7accee036662424cb8b9474a3c480bfb : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d8420fd20806cac351b3cd2c438c0a051763e68f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-28 15:10:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d41f49c4a63c5582e82b4a3057d466d629f47bbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 15:23:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/c4afbe3cae2f4553b078b7fa6a8c2571 : SUCCESS in 5m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c13a929017fb45ddaa33980556677f0c : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":2},{"id":"a6e17932a844a74e418332a5ccaa03cf4468a437","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-28 15:52:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d92353640cbaac08a3d8d7209a924500e2f8d406","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:09:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b8e43740dd5e4b73ad7e170639a93edc : SUCCESS in 6m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f16b900606e14a5785f3863ee6aef811 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"987f450d644752a6a7c3d60812e27efed4af1e8b","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-01 05:41:22.000000000","message":"Patch Set 3:\n\nLGTM, needs PTL +1.","accounts_in_message":[],"_revision_number":3},{"id":"64066f34eeb34ac4038b4dd51f16931a67a047c9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-01 06:07:42.000000000","message":"Patch Set 3:\n\nWe need to make them voting in tests repo, yes.","accounts_in_message":[],"_revision_number":3},{"id":"a72e940aba15818eedcc130f32b4da9255b6cd1a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-01 06:08:15.000000000","message":"Patch Set 3:\n\nDisregard please, wrong patch posted","accounts_in_message":[],"_revision_number":3},{"id":"1c050f9889b6ebde3c48a6ec9f71617dbdcf3bb6","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2020-09-01 13:12:31.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c4a5c1f8e8326b127ac23e0f2e4e8dd256839667","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-01 14:24:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"02f07d3272d1fca30b736b7aae5cc8ff9d52779d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:26:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f193587f10005f6a1cec44b845fb8530ec7e23cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:43:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0793c329a229423baa8e53f5042d6cbd : SUCCESS in 7m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2e98796d1d3a4c87a7b884ee7844cec5 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"922b0d4d933b54061251a404680ba13c7d0b2a35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 14:43:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ee1e239c272732c34f03b19a18eda327e6b8ee98","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 15:12:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f32e9a8d373d4a03928326e49411876b : SUCCESS in 3m 03s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/95e6a10e240b4320889c9af64d97984d : SUCCESS in 16m 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ba3c831d6a876cccea2593d1215ee6f7c3b7475b","revisions":{"8ae6062dafb8080419025471484ad58694ae9366":{"kind":"REWORK","_number":1,"created":"2020-08-28 14:53:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/748683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/748683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/748683/1"}}},"commit":{"parents":[{"commit":"d216a051dd87f3e4ddd93cd576e7909532d30ea9","subject":"Merge \"Remove legacy-tempest-dsvm-neutron-{ipv6only,serviceipv6}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d216a051dd87f3e4ddd93cd576e7909532d30ea9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-28 14:53:16.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-28 14:53:16.000000000","tz":180},"subject":"Add openstack-ansible/os_senlin role","message":"Add openstack-ansible/os_senlin role\n\nWe\u0027re going to implement deployment role of the senlin project\nas part of the OpenStack-Ansible tooling.\n\nChange-Id: I46c63250c4081c9915dc83b61880ad88f63f8efc\nNeeded-By: https://review.opendev.org/748677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ae6062dafb8080419025471484ad58694ae9366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8ae6062dafb8080419025471484ad58694ae9366"}]},"branch":"refs/heads/master"},"0f17795e1b00eb10a7a6fafcecfe0435474cfb05":{"kind":"REWORK","_number":2,"created":"2020-08-28 15:10:27.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/748683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/748683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/748683/2"}}},"commit":{"parents":[{"commit":"d216a051dd87f3e4ddd93cd576e7909532d30ea9","subject":"Merge \"Remove legacy-tempest-dsvm-neutron-{ipv6only,serviceipv6}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d216a051dd87f3e4ddd93cd576e7909532d30ea9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-28 14:53:16.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-28 15:10:09.000000000","tz":180},"subject":"Add openstack-ansible/os_senlin role","message":"Add openstack-ansible/os_senlin role\n\nWe\u0027re going to implement deployment role of the senlin project\nas part of the OpenStack-Ansible tooling.\n\nChange-Id: I46c63250c4081c9915dc83b61880ad88f63f8efc\nNeeded-By: https://review.opendev.org/748677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f17795e1b00eb10a7a6fafcecfe0435474cfb05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0f17795e1b00eb10a7a6fafcecfe0435474cfb05"}]},"branch":"refs/heads/master"},"ba3c831d6a876cccea2593d1215ee6f7c3b7475b":{"kind":"REWORK","_number":3,"created":"2020-08-28 15:52:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/83/748683/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/83/748683/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/83/748683/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/83/748683/3"}}},"commit":{"parents":[{"commit":"d216a051dd87f3e4ddd93cd576e7909532d30ea9","subject":"Merge \"Remove legacy-tempest-dsvm-neutron-{ipv6only,serviceipv6}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d216a051dd87f3e4ddd93cd576e7909532d30ea9"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-28 14:53:16.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-08-28 15:52:08.000000000","tz":180},"subject":"Add openstack-ansible/os_senlin role","message":"Add openstack-ansible/os_senlin role\n\nWe\u0027re going to implement deployment role of the senlin project\nas part of the OpenStack-Ansible tooling.\n\nChange-Id: I46c63250c4081c9915dc83b61880ad88f63f8efc\nNeeded-By: https://review.opendev.org/748677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba3c831d6a876cccea2593d1215ee6f7c3b7475b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ba3c831d6a876cccea2593d1215ee6f7c3b7475b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
