)]}'
{"id":"openstack%2Fproject-config~748712","triplet_id":"openstack%2Fproject-config~master~I5ca41e3d1d55f0ea54dc783d62f37903dcd99c2e","project":"openstack/project-config","branch":"master","topic":"retire-devstack-plugin-pika","hashtags":[],"change_id":"I5ca41e3d1d55f0ea54dc783d62f37903dcd99c2e","subject":"Retire devstack-plugin-pika project","status":"MERGED","created":"2020-08-28 16:07:48.000000000","updated":"2020-09-10 20:37:39.000000000","submitted":"2020-09-10 20:05:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748712-1599768325728-bbaa24f1","meta_rev_id":"efd7253a5e2e9c5f8f9f9fab7809c72c9fd228f8","_number":748712,"virtual_id_number":748712,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 20:05:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":2,"date":"2020-09-10 19:55:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-09-10 14:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":0,"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"value":1,"date":"2020-09-10 19:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 11:26:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-10 14:41:43.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-10 19:55:43.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-10 20:05:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a300ea9ff48d5da14148ebf2883f98893f955f99","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-28 16:07:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f6a6084ac60558f78e50c832a364a2de69cdd14","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-28 16:08:08.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d7e61a7823954da806c2110a42c0e005af30a4eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 16:23:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9098a8c3fe3a4b8a95f43ae08c53c88b : SUCCESS in 9m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/24fd83c900f0437db0f04ae2592ac3e1 : SUCCESS in 6m 57s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/f51b619d86404dc8ac89c707e1e7d0ea : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b949bc7e31f454fe1dbf16fec108ee80bca26763","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-28 16:24:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nLGTM,\n\nJust a minor concern inline","accounts_in_message":[],"_revision_number":1},{"id":"a36fe8ca21dc49b845dab1866b17c797a3ad25ec","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-28 18:05:40.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"926deaf42ed41ed7511972073644c3564ae4798e","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2020-08-28 18:08:29.000000000","message":"Patch Set 1:\n\nHi Hervé - IIUC Depends-On is used for cross-repo dependencies.  This change is in the same repo so I followed the procedure here instead:\n\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#adding-a-dependency","accounts_in_message":[],"_revision_number":1},{"id":"b366a2b3a4abdc1a12dd5c4bb79970bd097cab62","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-08-31 08:48:54.000000000","message":"Patch Set 1:\n\n\u003e Hi Hervé - IIUC Depends-On is used for cross-repo dependencies. \n \u003e This change is in the same repo so I followed the procedure here\n \u003e instead:\n \u003e \n \u003e https://docs.opendev.org/opendev/infra-manual/latest/developers.html#adding-a-dependency\n\nHi Ken,\n\nNo, the patch that I speak about is not in the same repo, it\u0027s against the openstack/devstack-plugin-pika repo [1], so I think this is a cross-dependency, but it\u0027s not an earth quake either and we can continue without that, we just need to manage of the correct ordering of the merges.\n\nhttps://review.opendev.org/#/c/748698/","accounts_in_message":[],"_revision_number":1},{"id":"2f6f7994033843df3e9fdf5ba47e0a943ba14122","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-08-31 18:30:05.000000000","message":"Patch Set 1:\n\nThis should indeed have a depends-on the pika change to not merge it too early.","accounts_in_message":[],"_revision_number":1},{"id":"8aa170cd121a45469499604ed72b07e77912c18b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-03 20:57:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"138f3ad8d6f891e3d709d5336a39a6acb778875e","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-03 20:57:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"41befa2764a7bd18ee9ff4c01e0f6eeeb8dd71ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:24:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9608e4c3507841a7bbbb8c2fd4a57ef6 : SUCCESS in 6m 58s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3fb93650582649c5a468eff28eaad7b3 : SUCCESS in 5m 17s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/795ed3d8b99c4d288f3d465c894156d8 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":2},{"id":"5d08c71108b386dad54b7f7482b07fecc0a3015e","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-04 10:31:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3979e7c0df59ee00cf73867cab1fe2bbfeed1c46","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-04 10:32:18.000000000","message":"Patch Set 2: -Workflow\n\nwaiting for depends-on to merge","accounts_in_message":[],"_revision_number":2},{"id":"1d3e03b0b34e3432f7536c1d84ea90a0a875e34e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 11:26:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ad1902d3b00c40f76de144d41cfab9a5cb09d4de","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 18:47:54.000000000","message":"Patch Set 2: Workflow+1\n\ndependency merged","accounts_in_message":[],"_revision_number":2},{"id":"b78f996fdba337498ad9da672222c8a0c7f12b34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 18:48:11.000000000","message":"Patch Set 2: Verified-2\n\n(3 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/project-config on branch master.  The error was:\n\n  Unknown projects: openstack/networking-l2gw\n\nThe error appears in the following project stanza:\n\n  project:\n      name: openstack/networking-midonet\n      templates:\n        - official-openstack-repo-jobs\n        - periodic-jobs-with-neutron-lib-master\n        - translation-jobs-master-only\n        - publish-to-pypi\n      periodic:\n        jobs:\n          - openstack-tox-py36-with-neutron-lib-master:\n              required-projects:\n                - openstack/neutron-fwaas\n                - openstack/neutron-vpnaas\n                - openstack/neutron-dynamic-routing\n                - openstack/networking-l2gw\n                - x/tap-as-a-service\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 3361, column 3","accounts_in_message":[],"_revision_number":2},{"id":"a1dc285ff0d70eac47f92d4317b4ab9a75db6c79","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 19:01:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dc96b83d46af732205ce65fef25b48835edb141c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:25:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/976aa0871002442fbd7c7ef74f07c4dc : SUCCESS in 6m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a6a8b13d39424943a4829f60af16a07b : SUCCESS in 4m 58s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/db6852125bfe4e2fa72dfc30ef52528a : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bb9a54e721463b5a49154f4385a22a850a0c70f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:36:19.000000000","message":"Patch Set 2: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"18b38177d4ede2164403032795ee58aee38dd69c","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-10 14:40:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d1a9c3dce4ba63278d7296084e6e502931618dad","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-10 14:41:43.000000000","message":"Patch Set 3: Code-Review+2\n\nrebased and resolved merge conflicts","accounts_in_message":[],"_revision_number":3},{"id":"3817776a5723c7462f86f8dee1125761811ec5df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 14:55:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/af86f354192a4c708e6d731eb148e91d : SUCCESS in 6m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9c322689dc624b9b8c2f0fc87b9d1ad1 : SUCCESS in 4m 35s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/20d15a4eedc04b629e7ba501e53d0b30 : SUCCESS in 8m 05s","accounts_in_message":[],"_revision_number":3},{"id":"62433f79795a2a057c939af5a5130d9862ef7265","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-10 19:55:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c645ade9d35f75a65d0b14320a8dbeae1acb5ba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 19:56:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cc19a435f0353ffd7cb6f208d99edbba22cfe42d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:05:25.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d1ee3654827b4f49abc73299edab4e0a : SUCCESS in 6m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/115dc1ca82cf42cb87f55c0dd01be622 : SUCCESS in 4m 36s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/330013b4fc3c497e8b707511d7c995bd : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"253244b9d8b32f8114fd53ef2ea014d000eb2e86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:05:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"a9365c8c28884a0550965842b5a4f3153c4963b6","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:37:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/1136a027c1e440f6bd6fa9dcdf457580 : SUCCESS in 2m 14s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/f33728250a5d427ea7d65e77bddfbbd1 : SUCCESS in 10m 37s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/67c04fbb750e4d58a4784d2f414a63ea : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"749af9f052fbcbb501bab745aaf9e09b4212939c","revisions":{"d9f52ca66ec677fde19579b7a0513cecbb9db2a3":{"kind":"REWORK","_number":1,"created":"2020-08-28 16:07:48.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/12/748712/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/748712/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/748712/1"}}},"commit":{"parents":[{"commit":"9e1604b5103302eb8a0d357cf3c21b0f55209831","subject":"Retire devstack-plugin-pika project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e1604b5103302eb8a0d357cf3c21b0f55209831"}]}],"author":{"name":"Ken Giusti","email":"kgiusti@gmail.com","date":"2020-08-28 16:05:56.000000000","tz":-240},"committer":{"name":"Ken Giusti","email":"kgiusti@gmail.com","date":"2020-08-28 16:07:24.000000000","tz":-240},"subject":"Retire devstack-plugin-pika project","message":"Retire devstack-plugin-pika project\n\nThis plugin was used by the oslo.messaging project to test the\ndevelopment of an alternative RabbitMQ message driver based on\nthe Pika library (instead of the default Kombu driver).\n\nDevelopment of this driver was dropped in Queens and is not\nsupported.\n\nChange-Id: I5ca41e3d1d55f0ea54dc783d62f37903dcd99c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9f52ca66ec677fde19579b7a0513cecbb9db2a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d9f52ca66ec677fde19579b7a0513cecbb9db2a3"}]},"branch":"refs/heads/master"},"f09bb026463fc418158fc121b2104ac13295f2ff":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-03 20:57:23.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/12/748712/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/748712/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/748712/2"}}},"commit":{"parents":[{"commit":"9e1604b5103302eb8a0d357cf3c21b0f55209831","subject":"Retire devstack-plugin-pika project","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9e1604b5103302eb8a0d357cf3c21b0f55209831"}]}],"author":{"name":"Ken Giusti","email":"kgiusti@gmail.com","date":"2020-08-28 16:05:56.000000000","tz":-240},"committer":{"name":"Clark Boylan","email":"cboylan@sapwetik.org","date":"2020-09-03 20:57:20.000000000","tz":0},"subject":"Retire devstack-plugin-pika project","message":"Retire devstack-plugin-pika project\n\nThis plugin was used by the oslo.messaging project to test the\ndevelopment of an alternative RabbitMQ message driver based on\nthe Pika library (instead of the default Kombu driver).\n\nDevelopment of this driver was dropped in Queens and is not\nsupported.\n\nDepends-On: https://review.opendev.org/#/c/748698/\nChange-Id: I5ca41e3d1d55f0ea54dc783d62f37903dcd99c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f09bb026463fc418158fc121b2104ac13295f2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f09bb026463fc418158fc121b2104ac13295f2ff"}]},"branch":"refs/heads/master"},"749af9f052fbcbb501bab745aaf9e09b4212939c":{"kind":"REWORK","_number":3,"created":"2020-09-10 14:40:39.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/12/748712/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/12/748712/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/12/748712/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/12/748712/3"}}},"commit":{"parents":[{"commit":"9f9b47a8421d8e114907395354dd9c9c5c0b854a","subject":"Merge \"pynotedb: remove project from infrastructure systems\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/9f9b47a8421d8e114907395354dd9c9c5c0b854a"}]}],"author":{"name":"Ken Giusti","email":"kgiusti@gmail.com","date":"2020-08-28 16:05:56.000000000","tz":-240},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-09-10 14:40:10.000000000","tz":120},"subject":"Retire devstack-plugin-pika project","message":"Retire devstack-plugin-pika project\n\nThis plugin was used by the oslo.messaging project to test the\ndevelopment of an alternative RabbitMQ message driver based on\nthe Pika library (instead of the default Kombu driver).\n\nDevelopment of this driver was dropped in Queens and is not\nsupported.\n\nDepends-On: https://review.opendev.org/#/c/748698/\nChange-Id: I5ca41e3d1d55f0ea54dc783d62f37903dcd99c2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/749af9f052fbcbb501bab745aaf9e09b4212939c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/749af9f052fbcbb501bab745aaf9e09b4212939c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
