)]}'
{"id":"openstack%2Fproject-config~749834","triplet_id":"openstack%2Fproject-config~master~I1d2c40a0b07e7e0816d4381253655bd05ff2ab79","project":"openstack/project-config","branch":"master","topic":"new-project","hashtags":[],"change_id":"I1d2c40a0b07e7e0816d4381253655bd05ff2ab79","subject":"Add openstack/osops","status":"MERGED","created":"2020-09-03 21:39:54.000000000","updated":"2020-09-11 00:47:45.000000000","submitted":"2020-09-10 20:18:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749834-1599769081105-b51fe808","meta_rev_id":"69cf2ba6613d3d7f9830fb7d41f163fa5c59c6f3","_number":749834,"virtual_id_number":749834,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 20:18:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-10 19:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-09-07 19:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-09-04 11:29:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-10 09:59:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2020-09-10 19:57:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 11:29:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-07 19:38:58.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-10 09:59:50.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2020-09-10 19:57:36.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-10 20:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f680482f16e66e938c750a71b5426afb47b10275","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-03 21:39:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88bf18166a09ebba28b30ba99f987740be651f55","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-03 21:40:45.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7b3fef7cc4c133302e81e8c63ba486955639aff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 21:55:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e52ef09616b46eb808da5a33dbd229b : SUCCESS in 5m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/74fa58191df8470da225854aec8812fa : SUCCESS in 4m 57s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/74ec5e5fee1e4621b2ee2d480eb1e063 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":2},{"id":"480d4db09762643de8b073b5bf62093ccdbc4029","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-04 11:29:56.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"665fcd4fbbf04f01996cf2dabda6fbc4a6d7dce9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-07 19:38:58.000000000","message":"Patch Set 2: Code-Review+2\n\nto sync this to github, please send a followup change to add the official-openstack-repo-jobs job template.","accounts_in_message":[],"_revision_number":2},{"id":"b191d2105830649f54f43479c2d0cc720d3f7a96","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2020-09-10 09:59:50.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"40014c74427ed264149f794557bf1577e348faeb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-10 19:57:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af4a347adb974c44af0a82356dae56dae6842ab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 19:57:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b3523decc4af4594df15dbf5aaac432c0ac4c59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:18:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee8f212b0d154e439855151b5943a665 : SUCCESS in 6m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9964e7183e2c4cc5a529306724f1225f : SUCCESS in 4m 41s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/9998e83f97e944f5ae0ee1143020e441 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":2},{"id":"2d97d06c6a165d5ec48ccbe989bb444bab825854","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 20:18:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c80552d9bf578e8c45d4aebd8a830093b287961b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 21:03:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/76d15d84552941cc89ddb2aff78b00a7 : SUCCESS in 3m 30s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6ded1d3d44ae4a68829683a9140bb49e : SUCCESS in 10m 30s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/d1e350ed9e064aa0bedf64c7c8edc181 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"018bcc4c893d93009c319aa0b5bb24dca5f90aa1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 00:47:45.000000000","message":"Patch Set 2:\n\n\u003e to sync this to github, please send a followup change to add the\n \u003e official-openstack-repo-jobs job template.\n\nFollow up posted with https://review.opendev.org/751124","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53e2f78149da635d179f6322e9827d13e33c7c82","revisions":{"d1c0ec53305f3c309476236733fe2867550c1547":{"kind":"REWORK","_number":1,"created":"2020-09-03 21:39:54.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/749834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/749834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/749834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/749834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/749834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/749834/1"}}},"commit":{"parents":[{"commit":"7cf1a70cb6525e084f510ff720398abd454166fb","subject":"Merge \"kolla-cli: deprecation - remove noop jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7cf1a70cb6525e084f510ff720398abd454166fb"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 21:36:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 21:38:59.000000000","tz":-300},"subject":"Add openstack/osops","message":"Add openstack/osops\n\nThis adds a new repo to combine the several x/osops-* repos into one\nrepo owned by the Operation Docs and Tooling SIG as a place to collect\nscripts, code snippets, configs, and any other useful tools for\noperators.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-August/016858.html\n\nChange-Id: I1d2c40a0b07e7e0816d4381253655bd05ff2ab79\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1c0ec53305f3c309476236733fe2867550c1547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d1c0ec53305f3c309476236733fe2867550c1547"}]},"branch":"refs/heads/master"},"53e2f78149da635d179f6322e9827d13e33c7c82":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-03 21:40:45.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/34/749834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/34/749834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/749834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/749834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/34/749834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/34/749834/2"}}},"commit":{"parents":[{"commit":"7cf1a70cb6525e084f510ff720398abd454166fb","subject":"Merge \"kolla-cli: deprecation - remove noop jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7cf1a70cb6525e084f510ff720398abd454166fb"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 21:36:41.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-09-03 21:40:24.000000000","tz":-300},"subject":"Add openstack/osops","message":"Add openstack/osops\n\nThis adds a new repo to combine the several x/osops-* repos into one\nrepo owned by the Operation Docs and Tooling SIG as a place to collect\nscripts, code snippets, configs, and any other useful tools for\noperators.\n\nhttp://lists.openstack.org/pipermail/openstack-discuss/2020-August/016858.html\n\nNeeded-by: https://review.opendev.org/749835\n\nChange-Id: I1d2c40a0b07e7e0816d4381253655bd05ff2ab79\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53e2f78149da635d179f6322e9827d13e33c7c82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/53e2f78149da635d179f6322e9827d13e33c7c82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
