)]}'
{"id":"openstack%2Fproject-config~749993","triplet_id":"openstack%2Fproject-config~master~I96f69bfa7e3bb2928f24755fd1ff2cb1b347548c","project":"openstack/project-config","branch":"master","topic":"mirror-reboot","hashtags":[],"change_id":"I96f69bfa7e3bb2928f24755fd1ff2cb1b347548c","subject":"Temporarily disable rax-dfw for mirror reboot","status":"MERGED","created":"2020-09-04 16:24:36.000000000","updated":"2020-09-04 17:27:55.000000000","submitted":"2020-09-04 17:12:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749993-1599239561374-42a114cc","meta_rev_id":"0a3b79cfc747263de8bcbc51fb40053d3964080b","_number":749993,"virtual_id_number":749993,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-04 17:12:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2020-09-04 16:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-09-04 17:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2020-09-04 17:01:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 16:25:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-04 17:01:47.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-04 17:12:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b9b45527200ba449900500e3cfad71b5e3675f42","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-09-04 16:24:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9b5af6d30c7cd2c228cefdab8d771f21cb26a38","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-04 16:25:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63620f6c26a6fa1b308c07cd71cc7d3e5478ec18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:39:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9cd3af6f8a154f528767744ef767092a : SUCCESS in 6m 50s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/073c324b687c4be1a004d736280c74d6 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1bcd9465c4111fde1f2e3079f0551a653a969e46","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-04 17:01:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0093d4db513552e8ca26fb71817bdd14048c8190","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:02:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be6470e13ee87b2b5936dbe315c51df3b8c359e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:12:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5acc32b891a24aa9a8286f8a75bc5b44 : SUCCESS in 6m 27s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/6899702dd7ef4b75a562b85fd98111b1 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"5fa3974a5bbf74378c6c0b911662053addfc4c0b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:12:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a6ff49051f41519e8a2bc77e36cc2bcaf25fae84","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 17:27:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/fab5d92c7fec402c87195ab693a070e1 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c2a4a34a9ebfc9c6f0dd61648c412dab9165f55","revisions":{"3c2a4a34a9ebfc9c6f0dd61648c412dab9165f55":{"kind":"REWORK","_number":1,"created":"2020-09-04 16:24:36.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/93/749993/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/93/749993/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/749993/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/749993/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/93/749993/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/93/749993/1"}}},"commit":{"parents":[{"commit":"56d77522bf17165fa32e047b7c46d550fd498e2b","subject":"Merge \"Final step for networking-l2gw and networking-l2gw-tempest-plugin retirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56d77522bf17165fa32e047b7c46d550fd498e2b"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-04 16:21:45.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2020-09-04 16:21:45.000000000","tz":0},"subject":"Temporarily disable rax-dfw for mirror reboot","message":"Temporarily disable rax-dfw for mirror reboot\n\nWhile attempting to replace the Cinder volume attached to the\nrax-dfw mirror server, the xenwatch reported a page allocation\nfailure. In order to ensure the new block device is correctly seen\nby the guest OS we should reboot it. Temporarily wind down testing\nthere so as not to disrupt any running jobs when we do so. This can\nbe reverted after we complete the reboot.\n\nChange-Id: I96f69bfa7e3bb2928f24755fd1ff2cb1b347548c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c2a4a34a9ebfc9c6f0dd61648c412dab9165f55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3c2a4a34a9ebfc9c6f0dd61648c412dab9165f55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
