)]}'
{"id":"openstack%2Fproject-config~749995","triplet_id":"openstack%2Fproject-config~master~If73b64e64bf23840c2b00a61a55879fea24a291f","project":"openstack/project-config","branch":"master","topic":"permit-copyAnyScore","hashtags":[],"change_id":"If73b64e64bf23840c2b00a61a55879fea24a291f","subject":"Allow copyAnyScore in gerrit ACLs","status":"ABANDONED","created":"2020-09-04 16:31:17.000000000","updated":"2020-09-04 16:45:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"64b0831143809489e128fcefd29f4767f03abdf0","_number":749995,"virtual_id_number":749995,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"date":"2020-09-04 16:38:39.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-04 16:42:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":-1,"date":"2020-09-04 16:33:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-04 16:33:35.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-04 16:42:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e774b3154438468782d07a88b7ab590d9c427e12","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-09-04 16:31:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7473508e664d75922fab78537119e122e5d561e6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-04 16:33:35.000000000","message":"Patch Set 1: Code-Review-1\n\nThis is not a valid function in our gerrit: https://review.opendev.org/Documentation/config-labels.html. When we get upgraded (slowly making progress on that this will change).","accounts_in_message":[],"_revision_number":1},{"id":"a6e8fbac87966b74c0a04535f04d2245c1063f9e","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-09-04 16:38:39.000000000","message":"Patch Set 1:\n\n\u003e This is not a valid function in our gerrit: https://review.opendev.org/Documentation/config-labels.html.\n \u003e When we get upgraded (slowly making progress on that this will\n \u003e change).\n\nAh, well shoot.  Is there a different method for not having a label be overwritten by subsequent patchsets in our Gerrit?","accounts_in_message":[],"_revision_number":1},{"id":"7132daf3457be60378f4e79e9ce110873b66365b","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-04 16:40:56.000000000","message":"Patch Set 1:\n\n\u003e \u003e This is not a valid function in our gerrit: https://review.opendev.org/Documentation/config-labels.html.\n \u003e \u003e When we get upgraded (slowly making progress on that this will\n \u003e \u003e change).\n \u003e \n \u003e Ah, well shoot.  Is there a different method for not having a label\n \u003e be overwritten by subsequent patchsets in our Gerrit?\n\nI don\u0027t think so.","accounts_in_message":[],"_revision_number":1},{"id":"086df85db6d5f109729e2456c7a6ed96d3d21e04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 16:42:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a10b7dff9c584984a979d42fbc3f0e68 : SUCCESS in 8m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8d7d823f3eb98a13da6cc263eeef983158b14631","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-09-04 16:45:26.000000000","message":"Abandoned\n\nI can revive this later perhaps, once our Gerrit supports it.  Thanks Clark!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"06ef650337bb800d3dcd777728e6caab927e8061","revisions":{"06ef650337bb800d3dcd777728e6caab927e8061":{"kind":"REWORK","_number":1,"created":"2020-09-04 16:31:17.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/95/749995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/749995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/749995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/749995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/749995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/749995/1"}}},"commit":{"parents":[{"commit":"56d77522bf17165fa32e047b7c46d550fd498e2b","subject":"Merge \"Final step for networking-l2gw and networking-l2gw-tempest-plugin retirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/56d77522bf17165fa32e047b7c46d550fd498e2b"}]}],"author":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-09-04 16:29:19.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"nate.johnston@redhat.com","date":"2020-09-04 16:31:15.000000000","tz":-240},"subject":"Allow copyAnyScore in gerrit ACLs","message":"Allow copyAnyScore in gerrit ACLs\n\nAccording to [1] copyAnyScore is a permissible ACL label option.\nPermit it\u0027s usage in Gerrit ACLs.\n\n[1] https://gerrit-review.googlesource.com/Documentation/config-labels.html#label_copyAnyScore\n\nChange-Id: If73b64e64bf23840c2b00a61a55879fea24a291f\nNeeded-By: https://review.opendev.org/749986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/06ef650337bb800d3dcd777728e6caab927e8061"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/06ef650337bb800d3dcd777728e6caab927e8061"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
