)]}'
{"id":"openstack%2Fproject-config~750635","triplet_id":"openstack%2Fproject-config~master~I16dd3a39d571458b3a7270d53620388cf18a14a0","project":"openstack/project-config","branch":"master","topic":"infra-retire","hashtags":[],"change_id":"I16dd3a39d571458b3a7270d53620388cf18a14a0","subject":"pynotedb: remove project from infrastructure systems","status":"MERGED","created":"2020-09-09 12:17:14.000000000","updated":"2020-09-10 09:07:50.000000000","submitted":"2020-09-10 06:58:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750635-1599721126027-b51fe808","meta_rev_id":"1c254086b2ad8d24e0965a5d7e72789f78cccd67","_number":750635,"virtual_id_number":750635,"owner":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 06:58:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2020-09-09 19:04:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-10 06:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-10 06:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 19:03:48.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-09-09 19:04:54.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-09-10 06:46:42.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-09-10 06:58:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67bca83129f09c48c7244cecbf40e75c37a05709","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2020-09-09 12:17:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"957253fb589a2b992cd0c2326624e989fc169836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 12:17:44.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":1},{"id":"ba0c357314bd0e5da10ea8e945500b417095f6e3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 19:03:19.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"571f6c186a824b44e25a86295ac8f3e593ec130f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-09-09 19:03:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3378c6df87fd3cb280985689b1f33b59f1e322e6","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 19:04:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nlet me fix this quickly","accounts_in_message":[],"_revision_number":2},{"id":"792b72516f0767ad663b7abeefea2396056318f0","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 19:04:50.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0c9115f0fbd04539f995e9b3bd913617c1188ee7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-09-09 19:04:54.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4517e3a820650ceac05b84b1d79f3d61e90f3d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 19:29:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7bbb058ab02b4e35b74179ffc8c7d2cc : SUCCESS in 8m 33s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c5792e62f1fc4861b5528e54b6e5708f : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":3},{"id":"190ba801dbae4b303c1521ded06c22aa20ddaa55","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-09-10 06:46:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"521b5e3586d054cfcf03513f6fbbc23ab63c7080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:47:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d14fb7915f52c213d2f37bc8ff334008ca30a487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:58:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/06e7b751e264427bab2bc73b3e0a5934 : SUCCESS in 6m 38s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8d39274e720649a68d53972965f6d715 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8639cd1529abba36f7b98874e3bad1f6c7b386ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 06:58:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e7872bf06302e0396c91d1e54575241dfe237fa0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 09:07:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/51dddb203cee453e83741f03c3028248 : SUCCESS in 2m 38s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/67079f0fac8645d1a78e062aed0e6290 : SUCCESS in 13m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3dae15b19e8c26358ec2f869a20950717d47ba1f","revisions":{"1f5adc3cd45e60fe91ff88e39d180e2a7d5755bc":{"kind":"REWORK","_number":1,"created":"2020-09-09 12:17:14.000000000","uploader":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"ref":"refs/changes/35/750635/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/750635/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/750635/1"}}},"commit":{"parents":[{"commit":"08d92637cf0695e447be0cf67d63878876a7f21a","subject":"pynotedb: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/08d92637cf0695e447be0cf67d63878876a7f21a"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2020-09-09 12:14:53.000000000","tz":0},"committer":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2020-09-09 12:17:01.000000000","tz":0},"subject":"pynotedb: remove project from infrastructure systems","message":"pynotedb: remove project from infrastructure systems\n\nThis project is retired.\n\nChange-Id: I16dd3a39d571458b3a7270d53620388cf18a14a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1f5adc3cd45e60fe91ff88e39d180e2a7d5755bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/1f5adc3cd45e60fe91ff88e39d180e2a7d5755bc"}]},"branch":"refs/heads/master"},"3fb1893bf971e076b4e579a39bef0aa7e99f16cd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-09 19:03:19.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/35/750635/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/750635/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/750635/2"}}},"commit":{"parents":[{"commit":"08d92637cf0695e447be0cf67d63878876a7f21a","subject":"pynotedb: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/08d92637cf0695e447be0cf67d63878876a7f21a"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2020-09-09 12:14:53.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-09-09 19:03:15.000000000","tz":0},"subject":"pynotedb: remove project from infrastructure systems","message":"pynotedb: remove project from infrastructure systems\n\nThis project is retired.\n\nDepends-On: https://review.opendev.org/597402\nChange-Id: I16dd3a39d571458b3a7270d53620388cf18a14a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3fb1893bf971e076b4e579a39bef0aa7e99f16cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3fb1893bf971e076b4e579a39bef0aa7e99f16cd"}]},"branch":"refs/heads/master"},"3dae15b19e8c26358ec2f869a20950717d47ba1f":{"kind":"REWORK","_number":3,"created":"2020-09-09 19:04:50.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/35/750635/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/35/750635/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/35/750635/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/35/750635/3"}}},"commit":{"parents":[{"commit":"08d92637cf0695e447be0cf67d63878876a7f21a","subject":"pynotedb: end project gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/08d92637cf0695e447be0cf67d63878876a7f21a"}]}],"author":{"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","date":"2020-09-09 12:14:53.000000000","tz":0},"committer":{"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","date":"2020-09-09 19:04:47.000000000","tz":0},"subject":"pynotedb: remove project from infrastructure systems","message":"pynotedb: remove project from infrastructure systems\n\nThis project is retired.\n\nDepends-On: https://review.opendev.org/597402\nChange-Id: I16dd3a39d571458b3a7270d53620388cf18a14a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dae15b19e8c26358ec2f869a20950717d47ba1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3dae15b19e8c26358ec2f869a20950717d47ba1f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
