)]}'
{"id":"openstack%2Fproject-config~750642","triplet_id":"openstack%2Fproject-config~master~I0acf6c581cd22a8835928ef643d4237f66ca6181","project":"openstack/project-config","branch":"master","topic":"f32","hashtags":[],"change_id":"I0acf6c581cd22a8835928ef643d4237f66ca6181","subject":"Add Fedora 32 builds","status":"MERGED","created":"2020-09-09 12:36:21.000000000","updated":"2020-09-10 04:24:18.000000000","submitted":"2020-09-10 04:06:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750642-1599710806837-210ac784","meta_rev_id":"cbd38167e25afab021cc7aaea93e5b9150e8e9fa","_number":750642,"virtual_id_number":750642,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-10 04:06:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-10 03:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-10 03:46:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-09 14:23:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-09-10 03:46:49.000000000","updated_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"},{"updated":"2020-09-10 04:06:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe73b5b881ed1f74ce3275e1873e871e5fc0d7b6","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-09 12:36:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01daeddffc32bcf0050f5415317972da5d81c02b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-09 12:39:01.000000000","message":"Topic set to fedora-32","accounts_in_message":[],"_revision_number":1},{"id":"808b5d3e87c663d79ab9c22fef6b5ac8a4aa589f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 12:47:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0bb6f165136a47d594923e15aeb73e4e : SUCCESS in 7m 26s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/892b3c8bd0ff4ee0ac395999a77fdd4b : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a89e6c5a0f833abb59496d59642be8a988b4b85c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-09 12:52:04.000000000","message":"Topic changed from fedora-32 to f32","accounts_in_message":[],"_revision_number":1},{"id":"3f222bf2be7b3c3dd993189c6efc1786d0f250fb","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-09 12:57:59.000000000","message":"Patch Set 1:\n\nWIP as I\u0027m not entirely sure if this is possible to do in one shot, previously for f31 it took a few changes to enable fully but I believe that was more due to issues building the image in certain envs?","accounts_in_message":[],"_revision_number":1},{"id":"e550e33cd681474d13b62effcd87a952f1c1ea06","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-09-09 14:23:37.000000000","message":"Patch Set 1:\n\nYou need to change the builder config first, see https://review.opendev.org/#/c/716127/\n\nAlso likely we want to keep the change to nodepool.yaml separate.","accounts_in_message":[],"_revision_number":1},{"id":"e6ada9f39b77bb32c7e814aea33f6c9aa36ce4ee","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-09-10 03:45:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"11fd5506e7a4cb69d2e63862a4d1bc3866ea7ab6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2020-09-10 03:46:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI don\u0027t think we\u0027ll have any issue building this so will approve as is, and watch the builds this afternoon","accounts_in_message":[],"_revision_number":2},{"id":"4ed9db85cc957869436cee28fec168b0d3332062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:57:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8431683060aa461c9da1925c223a3e45 : SUCCESS in 6m 08s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/67d01f6209c54b4f8b624ac62cfe3ba8 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"8084931ece38c158aa806558fa176389205c0eff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 03:58:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7f317059e88306aa156748776da59cf632dd4841","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 04:06:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"086843532c03171963f4b53daec7a38ee517c37f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 04:06:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4caa2482099045d98dee0dbc19d94d07 : SUCCESS in 7m 31s\n- project-config-nodepool https://zuul.opendev.org/t/openstack/build/9fb609b561734b9a82e3b987e89810ad : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":2},{"id":"cbd38167e25afab021cc7aaea93e5b9150e8e9fa","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 04:24:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\n\n- infra-prod-service-nodepool https://zuul.opendev.org/t/openstack/build/bc82b0bc5fe942f9a2cc0435e574189f : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"91da9f92f68f2a716d1eed88a969b07b128e1419","revisions":{"e9c805d3d678056eae89508ed25229dc5227beae":{"kind":"REWORK","_number":1,"created":"2020-09-09 12:36:21.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/42/750642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/750642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/750642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/750642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/750642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/750642/1"}}},"commit":{"parents":[{"commit":"425fb5e5c0deebc95885aa0cb3db7cef6fb7e082","subject":"Add monasca projects in vexxhost tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/425fb5e5c0deebc95885aa0cb3db7cef6fb7e082"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-09 12:34:55.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-09 12:34:55.000000000","tz":60},"subject":"WIP Add Fedora 32 builds","message":"WIP Add Fedora 32 builds\n\nChange-Id: I0acf6c581cd22a8835928ef643d4237f66ca6181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e9c805d3d678056eae89508ed25229dc5227beae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e9c805d3d678056eae89508ed25229dc5227beae"}]},"branch":"refs/heads/master"},"91da9f92f68f2a716d1eed88a969b07b128e1419":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-10 03:45:42.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/42/750642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/42/750642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/750642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/750642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/42/750642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/42/750642/2"}}},"commit":{"parents":[{"commit":"425fb5e5c0deebc95885aa0cb3db7cef6fb7e082","subject":"Add monasca projects in vexxhost tenant","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/425fb5e5c0deebc95885aa0cb3db7cef6fb7e082"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-09 12:34:55.000000000","tz":60},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2020-09-10 03:45:08.000000000","tz":600},"subject":"Add Fedora 32 builds","message":"Add Fedora 32 builds\n\nChange-Id: I0acf6c581cd22a8835928ef643d4237f66ca6181\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91da9f92f68f2a716d1eed88a969b07b128e1419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/91da9f92f68f2a716d1eed88a969b07b128e1419"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
